)]}'
{"id":"openstack%2Ffreezer~973623","triplet_id":"openstack%2Ffreezer~master~I2b4b4c3731a0a07e565caeab850dc340d95aeffa","project":"openstack/freezer","branch":"master","topic":"fix/nova-cinder-glance-integration","attention_set":{},"removed_from_attention_set":{"36848":{"account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"last_update":"2026-01-26 16:45:50.000000000","reason":"Change was submitted"},"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-01-26 16:45:50.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-21 09:00:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I2b4b4c3731a0a07e565caeab850dc340d95aeffa","subject":"Fix Glance integration","status":"MERGED","created":"2026-01-16 14:01:05.000000000","updated":"2026-01-26 16:47:13.000000000","submitted":"2026-01-26 16:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973623-fix/nova-cinder-glance-integration","meta_rev_id":"7448aca6b818c896d39ee6bfccc01772c90a5bd2","_number":973623,"virtual_id_number":973623,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 16:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-01-21 09:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-23 08:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-01-26 16:02:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 14:48:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 19:51:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-19 19:51:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"state":"REVIEWER"},{"updated":"2026-01-19 19:51:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-01-23 08:08:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e8eeb67ff78691b8cbf3d0b1dabae3d4a2973f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:01:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ac7c0cfb62adc99733f7677a16c229f86c0d943","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:01:23.000000000","message":"Topic fix/glance-integration removed","accounts_in_message":[],"_revision_number":1},{"id":"7a5af8aa914919c548b2d8a3052fccc6d036ecc6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:01:33.000000000","message":"Topic set to fix/nova-cinder-glance-integration","accounts_in_message":[],"_revision_number":1},{"id":"01e899cd3ac8d3d8beca3971a8ce936d54942359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:08:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64b8dd4b239913485de537635bf4e81cf1b1a2c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:10:47.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94b4aa4f2f3fa55e65cc0b4b847ef3981dd884dc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:19:18.000000000","message":"Topic fix/glance-integration removed","accounts_in_message":[],"_revision_number":3},{"id":"96527c23830af2fe9f3a00ebf7eb6b6a96525614","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:19:23.000000000","message":"Topic set to fix/nova-cinder-glance-integration","accounts_in_message":[],"_revision_number":3},{"id":"7f790ce4e4d6ea73526f2579300dbbe23a6999bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:48:49.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c005eaf343a4cd3833a2ec49e3c59ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499849e595cc4d708b491f435d415dd8 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba3c17362f7d4ca0b711182bda1bd8ba : FAILURE in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc8f0021ac874e68b5a2a6aac5e28243 : FAILURE in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f1366ec5b6044008636b743e7132cc1 : FAILURE in 4m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48a107b9dff14a03b14c84e3346f5482 : SUCCESS in 4m 08s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/cfb78a23cdf446f394b5fee4cc4f0e54 : SUCCESS in 34m 44s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e5f6b7490ee04266acbeae7e9ebdd263 : SUCCESS in 18m 50s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0d72d4d03c43ce8813f035d8208697 : SUCCESS in 31m 55s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/8ec8a12d565e4807b8ee3dce7b030aab : SUCCESS in 17m 51s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8c3aea7a2fd4a1ebc09ec63f95f7c7c : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":3},{"id":"fa63d36940a4c84892ae7ee050ae4eb661770e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 14:49:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_27665\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}],"_revision_number":4},{"id":"5f7562c93b2ca01a05f93d6a6eb2a09e21f1332c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 15:01:23.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"452bd7f7c3bfa29da9c75aa8a8ce7d91be3881b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 15:43:17.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5623256dc2e446e6bbfc11380a318acb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676838f0241442ce849a2a78c744151e : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5706b8a31dc8458f90801ebe8fee6315 : FAILURE in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fd82a06d29d4fe6b14a5ee132d2972c : FAILURE in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb73fdc65e4546c794c4fafd8443f56c : FAILURE in 3m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34e36ee8c59946b58977c8bd0cfd1e59 : SUCCESS in 4m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/7ade6b8feee24023a726cd0df98c21d6 : SUCCESS in 30m 36s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/bc8a526070614f8db87a80b2c4f11418 : SUCCESS in 37m 33s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be8e5f132983473eb18673d697a5e9a6 : SUCCESS in 36m 12s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/60bead56868845d4bf5ee3ff9f255e4d : SUCCESS in 35m 20s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3a7f8c4e9f94581a60cefd2c188b1f3 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":5},{"id":"1fddf5d93196d9fa4948a7094d0f5a520b530b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 16:25:43.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6fc70625677d7b1cfd60257c5ec865d0010f793a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:06:52.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/382856caa76c4f7ba03e05e3fdaa091a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650ec928993f4d068752f6dc32e454a0 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a23a2809885428b997339f67e55a0a8 : FAILURE in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32afe9ae5fdc4719ad06d1db6c2efea7 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d40ce459a924497c81dc5bd173024c4b : SUCCESS in 8m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e37110eb238f40e3acf44eb7c9df6d11 : SUCCESS in 3m 35s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/410f53d69311474582119b6fef7b71c7 : SUCCESS in 35m 06s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/4d5225814f704531bdee7afb4bf7a164 : SUCCESS in 27m 36s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31b6bfe141d54fe5ac1edd23bda0473d : SUCCESS in 32m 58s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/429f760dfedf41d98de5941e1fe01732 : SUCCESS in 21m 08s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/04fc7cf25a7048d6bd07f490a84fa884 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":6},{"id":"0fbcac524c5f89c7d4659decaedf65b44c580589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:13:22.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6656181cfc16438e194ed2818e6d1c0b346863b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:56:21.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15030137e6b446f5a6b38bc798368b17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909f44b3bc614091a70c425b840525e4 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430c6f051cbc4885b886aa1ac6a02f6d : FAILURE in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41fd1812c6f04bd7b1eddf4619b2fa68 : SUCCESS in 7m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5242c5532ac542d69f9fabeb85afdc86 : RETRY_LIMIT in 2m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a140927854dd49abab012aef9049c6e2 : SUCCESS in 2m 51s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ee091686aacb4297bca387f9174084e9 : SUCCESS in 33m 35s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/cd43214f90ed451c9411972ed706dd79 : SUCCESS in 35m 20s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/becd58b4489d4575b12c953c8b09cca2 : SUCCESS in 17m 08s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/712854a8c7774403bf9342b461046bf1 : SUCCESS in 18m 56s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e75189c023843a48fdef26479458a30 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":7},{"id":"7f8531d3857a210072e54e87d7439746dea68d9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 08:35:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4fa75d5680d62d4b80e7f078caab2809a19b4622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 08:57:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc95a1cc75ca47cabe0184e5313a9ed0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9dce8cb65341fa9f9bcd4e2caa4758 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d094cce2049484598147fec74f76163 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387e961cdcf54024bb154c98b1fbe4d1 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c708f68e5b284c61b27e612feef5596b : RETRY_LIMIT in 2m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc40cb4dcb794d5f8de19c6c0e43703f : SUCCESS in 4m 14s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/afda2a34d3cc423e91eb40d5befd17df : SUCCESS in 16m 30s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/0a9b0110a70f43d185028c00090827c9 : SUCCESS in 18m 58s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1a88dffacac4ce3ad4d799538ff4ca7 : SUCCESS in 15m 48s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/52dff6020b4447218f766c451388de1d : SUCCESS in 18m 34s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/00f9df74579b49d7b875dbb7058e1906 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":8},{"id":"0b5e029eb94795dc21608f5faf2f3d17f6ad5119","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 12:50:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0e1404c242498ceddc1deb948d1936e0317ba251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:37:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573eaa90d72a406da0e01d7b864db336\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e831e96ed3434ab6c4bd75d68a74c0 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc704fa9a37c41bbbd4a97acf425e283 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a884ef42419444b961bd969eeabc4a0 : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96ba7a507d1745c0a32e11729c37fde5 : SUCCESS in 3m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5dce68c038fb49b58625993cb1be065d : SUCCESS in 4m 48s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f1fd2cac42e442d79f3d31c2c66709af : SUCCESS in 37m 27s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/86f0094e9dfd48eb9407b0b1c865f77d : SUCCESS in 38m 56s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e501ddf9a1d4d74abecb8ea7d29187a : SUCCESS in 35m 26s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/e6df0c7261bd426286a955dfc5e7cfce : SUCCESS in 38m 02s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f8d6fda094e45de8114f0273302473a : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":8},{"id":"09c32ef3e806507573bb4959c1e7b11b7dbc8a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-20 16:33:41.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"c2639cc6d33355e19766a9f34d32ef0f1e775d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:17:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35908fcad2654710bb7d8b860359de06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d10ede6fcd245e49f88da52fb77bd65 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c6fe84babb2450cbe5a62a891d4c0de : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1eb1e5d92eb14964b8e6dee0acb34ea3 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ed6561288ac47c2a34d398f9d9075aa : SUCCESS in 3m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6cf25a00925d418ca1546a0305204dd9 : SUCCESS in 2m 17s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/de7b4c8c10224fefbfe055dddb1ad79d : SUCCESS in 17m 36s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/faba5b7c9cff4189bcdf3cb1a50729c2 : SUCCESS in 36m 54s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5599678ae2294324bb7cac5e48c200f7 : SUCCESS in 27m 26s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/f095662e956c4f568d3cb6f46678fee1 : SUCCESS in 18m 16s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/874951165a0f4ee2881861194374d81e : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":9},{"id":"443681f5316c9a440f2459ab2da74f2f16f93343","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-21 09:00:08.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"17fc6ffc52530fd03bca88e0899192b75a4a2d0b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 08:08:37.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"a794cb165949231f90712c0b845ee395fd8942ea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 16:02:02.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"d329b17e680270d3ab481000f57ef8e1c0cc3aeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:02:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"039c970eafc73017c7b392e4cbf3327c95385941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:45:49.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca088783dc934821a136db2e6f38c092\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80a98a75895430ea87dc2fcca21d668 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7017d6277fbc4079bf7e5f828c624b03 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2d2427384944ce3a4d1ea63a9fdc433 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cebd65add29346c6a2975a4899ff345d : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68c720dcc1684c91a459fd6957d73718 : SUCCESS in 4m 30s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e428f1457a714cd592f331ba62eed4da : SUCCESS in 36m 38s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/43f1898dcc614ec5a49f1c7546d15554 : SUCCESS in 25m 07s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b09ea6937654b07aeadd9e3517c4afd : SUCCESS in 29m 09s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/613c45c286ab4be790a389414b53f570 : SUCCESS in 37m 51s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/68ac6d3bc75e4ab980bb5d50eabf4a80 : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":9},{"id":"b0fd03c09435ebc99c2497bce4a1e2f8f050a5de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:45:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"7448aca6b818c896d39ee6bfccc01772c90a5bd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:47:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64249d34183d49fd8359b42602b2c970\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ed5e7a4c404686ad2af14d3f567ba9 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"981de1b6d435586e85220c544d689efd4b040eeb","revisions":{"c29d0b9d4a3585488349edbfd5cb74d89135d99f":{"kind":"REWORK","_number":1,"created":"2026-01-16 14:01:05.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/1"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:01:03.000000000","tz":60},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nDepends-On: 973620\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c29d0b9d4a3585488349edbfd5cb74d89135d99f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c29d0b9d4a3585488349edbfd5cb74d89135d99f"}]},"branch":"refs/heads/master"},"cae1b7affa52738eb3f3c723d8f9151e15ff12f0":{"kind":"REWORK","_number":2,"created":"2026-01-16 14:08:32.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/2"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:08:07.000000000","tz":60},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nDepends-On: 973620\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cae1b7affa52738eb3f3c723d8f9151e15ff12f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cae1b7affa52738eb3f3c723d8f9151e15ff12f0"}]},"branch":"refs/heads/master"},"d2cd3908279c5c1fa831d5a1aa49b9d5c1df2774":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-16 14:10:47.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/3"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:10:24.000000000","tz":60},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973620\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2cd3908279c5c1fa831d5a1aa49b9d5c1df2774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2cd3908279c5c1fa831d5a1aa49b9d5c1df2774"}]},"branch":"refs/heads/master"},"3448d7303dcf9b3c150cb54ed4ed1a7a0d7f7277":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-16 14:49:38.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/23/973623/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/4"}}},"commit":{"parents":[{"commit":"0c4cbcb3d3e4e74d623e7ed45122168a02739fca","subject":"Fix issues in Python code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0c4cbcb3d3e4e74d623e7ed45122168a02739fca"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:49:38.000000000","tz":0},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973620\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3448d7303dcf9b3c150cb54ed4ed1a7a0d7f7277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3448d7303dcf9b3c150cb54ed4ed1a7a0d7f7277"}]},"branch":"refs/heads/master","description":"Rebase"},"f30fd8bdbbf5394e7b1afaee6ee6db26e267f975":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-16 15:01:23.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/5"}}},"commit":{"parents":[{"commit":"8d57bc9bd120afd32b0b1e33b5faa7c7a08b807f","subject":"Fix issues in Python code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8d57bc9bd120afd32b0b1e33b5faa7c7a08b807f"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 15:00:56.000000000","tz":60},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f30fd8bdbbf5394e7b1afaee6ee6db26e267f975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f30fd8bdbbf5394e7b1afaee6ee6db26e267f975"}]},"branch":"refs/heads/master"},"5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-16 16:25:43.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/6"}}},"commit":{"parents":[{"commit":"e9c088782137f558cb566cf91e9899a2221509a5","subject":"Fix issues in Python code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e9c088782137f558cb566cf91e9899a2221509a5"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 16:25:43.000000000","tz":0},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa"}]},"branch":"refs/heads/master","description":"Rebase"},"3e1548caab593e1fddd3b29d656c4a09e2582d05":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-01-16 17:13:22.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/7"}}},"commit":{"parents":[{"commit":"74375ae6038dfe0be2dbe211fe75b68124c600c0","subject":"Fix issues in Python code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/74375ae6038dfe0be2dbe211fe75b68124c600c0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:13:22.000000000","tz":0},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3e1548caab593e1fddd3b29d656c4a09e2582d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3e1548caab593e1fddd3b29d656c4a09e2582d05"}]},"branch":"refs/heads/master","description":"Rebase"},"581f220a442dfcc2a12ffe458f64e27b87b8b3ef":{"kind":"REWORK","_number":8,"created":"2026-01-19 08:35:26.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/8"}}},"commit":{"parents":[{"commit":"74375ae6038dfe0be2dbe211fe75b68124c600c0","subject":"Fix issues in Python code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/74375ae6038dfe0be2dbe211fe75b68124c600c0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-19 08:35:11.000000000","tz":60},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/581f220a442dfcc2a12ffe458f64e27b87b8b3ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/581f220a442dfcc2a12ffe458f64e27b87b8b3ef"}]},"branch":"refs/heads/master"},"981de1b6d435586e85220c544d689efd4b040eeb":{"kind":"REWORK","_number":9,"created":"2026-01-20 16:33:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/23/973623/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/973623/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/973623/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/973623/9"}}},"commit":{"parents":[{"commit":"74375ae6038dfe0be2dbe211fe75b68124c600c0","subject":"Fix issues in Python code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/74375ae6038dfe0be2dbe211fe75b68124c600c0"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:49:13.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-20 16:33:17.000000000","tz":60},"subject":"Fix Glance integration","message":"Fix Glance integration\n\nChange-Id: I2b4b4c3731a0a07e565caeab850dc340d95aeffa\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/981de1b6d435586e85220c544d689efd4b040eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/981de1b6d435586e85220c544d689efd4b040eeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
