)]}'
{"id":"openstack%2Ffreezer~973625","triplet_id":"openstack%2Ffreezer~master~I9b2e1017d80695e1aea648e172c8def4dcff490d","project":"openstack/freezer","branch":"master","topic":"fix/nova-cinder-glance-integration","attention_set":{},"removed_from_attention_set":{"36848":{"account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"last_update":"2026-01-26 18:11:22.000000000","reason":"Change was submitted"},"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-01-26 18:11:22.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-26 16:30:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I9b2e1017d80695e1aea648e172c8def4dcff490d","subject":"Fix Nova integration and change clients to SDK","status":"MERGED","created":"2026-01-16 14:13:23.000000000","updated":"2026-01-26 18:12:14.000000000","submitted":"2026-01-26 18:11:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973625-fix/nova-cinder-glance-integration","meta_rev_id":"f7bafe8b6cce9862ab3dfd570ff00796bcca951e","_number":973625,"virtual_id_number":973625,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 18:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-01-26 16:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-01-26 16:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 14:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 13:12:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-01-19 19:51:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-19 19:51:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"state":"REVIEWER"},{"updated":"2026-01-19 19:51:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9322331c5a4d6ac558c445fcc2f36a530b3004e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e19707544b844b79dce7f80676c468d0866a478","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:13:35.000000000","message":"Topic fix/nova-integration removed","accounts_in_message":[],"_revision_number":1},{"id":"237acba9bdcbc56847c129eb27cec1db91854a26","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:13:44.000000000","message":"Topic set to fix/nova-cinder-glance-integration","accounts_in_message":[],"_revision_number":1},{"id":"e7291f3eb239b8c9f21f1e3391cce6e4d23aee68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:48:59.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/freezer/+/973620, https://review.opendev.org/c/openstack/freezer/+/973623 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"f5ef4bf08e688d82616687b1fcc79fcfa8cfb59b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:59:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"73489e708888316c342bf0c9b659d319b1584c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 15:01:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0d8d3ed1ca2d7bc83c2bbffa084463879e59acb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 15:44:57.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb88335a4625445ba0928177630ce7b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2322139aea4324bb6c9888b4390a6a : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d4a0c9b132427b9fb7fe66f42bd389 : FAILURE in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cc67e40716f4e088bcd2df8c9280aa9 : FAILURE in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ea17993ab04489ab4858b72f68be530 : FAILURE in 6m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c986b4203de842b09f52c5e07ae127e8 : SUCCESS in 4m 27s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b67ebea29ff94e109dfeff6cbe95e28d : SUCCESS in 19m 44s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a5e16ad64a7d46b2bf31615ebeeef1ef : SUCCESS in 33m 26s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bca98dcf89c4f93b5b6ad51963c166a : SUCCESS in 36m 38s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/4598a13527b74954b51bcd522d185a4e : SUCCESS in 37m 47s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d7451c22d844a49b705fca8a08411a7 : SUCCESS in 5m 25s\n\nWarning:\n  Comments left for invalid file freezer/engine/osbrick/client.py\n  Comments left for invalid file freezer/engine/osbrick/osbrick.py","accounts_in_message":[],"_revision_number":3},{"id":"367bca2a8430ff4ae242fffcf55171df71108a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 16:34:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8298b14cd27b925ba8cad0e49399db53153de128","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 16:39:11.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a2e7bc1ba3259ead464cd46889b26f34e3d4d2b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 16:58:42.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/123882cd9386450aadbcc1d3d59b2179\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10940d6f61b54df5a2dc07651f263e3b : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a9fa59f1fe54035918821cfe5cb08e2 : FAILURE in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7ca1fb307c744219b906db2f796fd6d : FAILURE in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/315b9eacbdd34fd891f9b10d1d277875 : RETRY_LIMIT in 1m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/792c1fc8d4a5496bae8341f13888f391 : SUCCESS in 2m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d4322067c3ef4f299f6ca0a202a89ecc : SUCCESS in 17m 20s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/65b5984a2e8a4df3b276a9af31ad689c : SUCCESS in 17m 36s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9eefb3423614fe69af38aa489dccc08 : SUCCESS in 17m 57s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/0ebc3de14cf54c77b7ce4373d40c71af : SUCCESS in 18m 02s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/01254cc005664cb9a93d9f119ee5a6ff : SUCCESS in 5m 34s\n\nWarning:\n  Comments left for invalid file freezer/engine/osbrick/client.py\n  Comments left for invalid file freezer/engine/osbrick/osbrick.py","accounts_in_message":[],"_revision_number":5},{"id":"6336c770b4baa821a61f4d4ad1b12b1379d7d5d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:16:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0b04a3764602ca13211132b38aec3441904efdf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:22:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"338e7134ce2e7f5ca86dcd3c31a6a450df409dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:05:06.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/471ad24dd7894a839649ac57ec381621\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433d8110bc2f445eb1e61cd8308929c6 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f727afe6c03442ebe194a02787abae3 : FAILURE in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/671efcd2c865456992647b8f008603e3 : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db90a989a0204cd6915ca240996f250f : RETRY_LIMIT in 2m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9cda62f7965b4f0bb014225b4796412e : SUCCESS in 2m 26s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/0cf04827e0154e93b7aeaf35a9f57d97 : SUCCESS in 30m 21s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b0117754cbae4371af1fd45668d969d2 : SUCCESS in 33m 39s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fadf8dccbcc402e945fa917d98907d3 : SUCCESS in 35m 54s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/21c4517bdbd148449f7662f5376aa761 : SUCCESS in 40m 17s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/9392f490758841e4a5c34348030b7156 : SUCCESS in 7m 46s\n\nWarning:\n  Comments left for invalid file freezer/engine/osbrick/client.py\n  Comments left for invalid file freezer/engine/osbrick/osbrick.py","accounts_in_message":[],"_revision_number":7},{"id":"3882a2b7f3534093e4c4f0d9c914708c89215f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 08:36:14.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"dd9f5707eba9c691c08c7210acac88f93913e614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 09:07:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff6d7bf666fb4b7295538c8be0038e5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d1c006961f498ca8e83a7461bf3f04 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f836797224184246a6ad2da5aece0ad9 : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b937afda3b67491096b8db6661d80e1b : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8f4673ef73240bea1a536fd2c3b780d : RETRY_LIMIT in 2m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ef31747f32a46a499213f28532d0d99 : SUCCESS in 3m 13s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/256d6129b91e469f9676b271f4d9b78a : SUCCESS in 27m 41s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/12fabc894e9d41ccbc7c3d026e91ff3a : SUCCESS in 16m 03s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bf2353f9a3f4cb49176c4e3dec217c3 : SUCCESS in 14m 52s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/2d2c1bc9b6504fc79914c46838749ad2 : SUCCESS in 21m 13s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7d981504ab34f1cbde11cf7197e27f5 : SUCCESS in 10m 35s\n\nWarning:\n  Comments left for invalid file freezer/engine/osbrick/client.py\n  Comments left for invalid file freezer/engine/osbrick/osbrick.py","accounts_in_message":[],"_revision_number":8},{"id":"c18f6fdcd6b99847264c022c0251590610c7afd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 09:50:41.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3a514636fc918e6838bfa9bbe4bd4ba334be23ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:35:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50940167567f4498ba1bd62880b34402\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c6f52c13f045b088233938bb8d8416 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da873864aff946d9a01ac62d443f8677 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8c2873c1c5c4118a51c2483135b898f : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca1a2a829b7141cab9b35284abceb5ff : RETRY_LIMIT in 1m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9febfafc4a04f0c9b74cf06bb6790cf : SUCCESS in 3m 29s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d9eff21de8f24a52a9e2d103e4295356 : SUCCESS in 37m 19s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e5a8ce0cc8f64bccafc85bbd71c1c9fc : SUCCESS in 35m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60da4f9793794a1a9d81a607a3767460 : SUCCESS in 35m 58s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/a4f01e015fba4610bda85ad34829990f : SUCCESS in 36m 29s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f343ee762db84f089fc33e4f7cfda732 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":9},{"id":"09f255f62fb87ea2c2917f7e1764e1c5933657f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 13:12:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e6cb68146f513e71d39b6e01d33b8e414d5541db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:06:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d3ef1c958bf40edb3eb176b5a1ba321\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf3897ae93045e79d0090a561ec3b99 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa71b8bd78e4502a2a063e8c88a1bc4 : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2f1873fe9df43bebe0ce63ce8d626c4 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f09e996044d947a29571b0fca5e84697 : SUCCESS in 3m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38e6869129dd42999ebd224d4e964c9c : SUCCESS in 3m 38s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ccde4ccd3dc84c01bda4cecef6d9e513 : SUCCESS in 34m 42s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8480088486c74d8799208cc4efcba0fb : SUCCESS in 33m 04s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/427a84bf6bad490c860374d8c2a278b7 : SUCCESS in 27m 42s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/744966a6a6384d82a21d28f7185d11d6 : SUCCESS in 44m 44s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/462374ff55794cd99b1085d2df714c6d : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":9},{"id":"9bb7d23f45386e80d5b3c9371bec6639e747f9ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-20 16:33:41.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"eb76164097022dad25ff19847fe26b1d8b6e1568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:22:41.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ebc96d6db3641fc99410548ac4db546\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7377a9f5c81e4624b31b138117033464 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd31a4becb344ca94cb99042a667c4b : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d53e577d7d61465e92e4d9a186da9527 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75dfdb11c650461992303b127513addb : SUCCESS in 6m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/184d34d9fce442bf8983a07ffb46c27b : SUCCESS in 2m 40s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/937c800695344f5792041ad5bcf35089 : SUCCESS in 17m 40s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/5a8d3c5feed2422bbb6e922c3627ef4e : SUCCESS in 15m 06s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a98cb8d933a94dd1b5fd5f89a16a32f6 : SUCCESS in 23m 43s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/100b448f5a814b9e9a0d93e2cc084790 : SUCCESS in 43m 18s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ca952577d484f0e8652b810e2ce36ad : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":10},{"id":"661a81660bee2d46afbf92f69394adc329b78e26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-22 14:04:29.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"75b1d3d00bd55e597710bd9d71c02f5a77eee6aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:06:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eac027411a6140958938452f3202e6f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84c4ecf4ccd4dafbe6975ea35af5151 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039912fdd3ee444ea74dc6e379eed829 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1a29b1cced840c6ba5722eeb05ba7b3 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee3218fe083a4a62921f98f753582bb8 : SUCCESS in 3m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a421427c605471380ad7291519f09dd : SUCCESS in 4m 37s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/24642e54bf13467da7318055efb2ff23 : SUCCESS in 35m 36s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8a2d69978bf148aab73219583558b197 : SUCCESS in 38m 44s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6a7df51969d45e8aaf3d0991e342c3d : FAILURE in 35m 58s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/aad29d3834a449a6a8bbaab3ffe6e3db : FAILURE in 37m 31s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/1819aa4315594874ae7abef7ef5a9a46 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":11},{"id":"ce9bd9e3754b439fc840f1ca8a18e036e1ecaf70","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 14:11:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a2deae5f35fd65b9223aeddda769401e7b946c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:04:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be003772021422a95cc8d55e5f795d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678bb09e92d341d89976957501ca66c9 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5d4d21f33b4fbe8da178c8a1c5359e : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/992e1b3a663842d4a55a2dbfe4dad147 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/beab699da2854ac29a87781b05011683 : SUCCESS in 4m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d4e23397156744aebc05708f6b451392 : SUCCESS in 4m 45s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/acf284adab0e4eb0a0e7ec0438082d54 : SUCCESS in 19m 32s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1426dab77b0c46e6926fc8b4caeed3e9 : SUCCESS in 34m 31s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9468013ae4ec42db9029cd80c7c69d2b : SUCCESS in 28m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/db11394ee6cb4c1a92a4bfce3898125a : SUCCESS in 18m 30s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/78492d473e7e4229acc4739a968e1c2c : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":11},{"id":"08bf5fa24049da39427aede832e6d9f305524e12","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 16:30:27.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"f0051a4e95386a9e2ff80280bd821c002a8ebb45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 16:58:40.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_27665\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}],"_revision_number":12},{"id":"18351cddd371e57199fbebba2f68d976e192427f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 16:58:53.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"b4f8e3c4d921d9013907f52c6913f627a8fe53b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 17:42:09.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dfe005ed48249feaacd552fbb35c599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70675758ce1242fab3731904e7a0660b : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9691ee95e48a4dde8f60140f68da10a8 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/512573d35a554f96a1cae904edf636f0 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8de623114b0e46faa59830973bdd0f68 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2071753e0c2c49838b4c4e005cdba8ad : SUCCESS in 4m 02s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/adf14170eb4541c3959f23f6ceb63e1b : SUCCESS in 35m 07s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/153ea8c40b494dc69e7b5d506b9c4c28 : SUCCESS in 24m 35s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80b11c52eeb64553a3d2fb1684fc13c9 : SUCCESS in 24m 20s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/5279e74a76664113af535b23860ef62c : SUCCESS in 38m 15s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bada8fb952b41c1bcaa4de3f34406d2 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":12},{"id":"f97cfcc3167439c341235056de0abddc23f85840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 17:42:40.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"30ed058b85b51fbf81b6a2403f031a987bb9d185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:11:22.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c8bf4b090d47b5ba96579f5514c773\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a13527366e40059bdb8c0de221c7a7 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009a2cd761684b5da9239611abc4a076 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e33db8d500db4ca283e78062267def84 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37169d9752254f859d47cafe4b738c72 : SUCCESS in 2m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ccecf37b362c4cca96806a0b567a8b79 : SUCCESS in 3m 50s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b3da149bd3e54f6c83af741cd49cb971 : SUCCESS in 27m 36s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/382ad371d81e464bbd73cc717f6f4de7 : SUCCESS in 17m 09s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffadedf187eb4866a616d782f30efe25 : SUCCESS in 21m 57s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/1d3d688475a6436f823e925fc9e1f02c : SUCCESS in 24m 52s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/25fb80e2547c44a484a318fc5b356c2d : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":12},{"id":"426882f2efb4f7a16fa8185f9fbf55c18d2de5d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:11:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"f7bafe8b6cce9862ab3dfd570ff00796bcca951e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:12:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90385488d6aa48aba86e5894907f7b97\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ad1bf17fd842cf8eed6c36dd7cbf97 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"d38e46be9440d93a98e01ab6a1f7155617e39c76","revisions":{"c6cad03e0ce038052af1bded1e026038b06a4858":{"kind":"REWORK","_number":1,"created":"2026-01-16 14:13:23.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/1"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:11:08.000000000","tz":60},"subject":"Fix Nova integration","message":"Fix Nova integration\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973620\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973623\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c6cad03e0ce038052af1bded1e026038b06a4858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c6cad03e0ce038052af1bded1e026038b06a4858"}]},"branch":"refs/heads/master"},"441c08a429f8aa245491c861dcf736aede8b30ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-16 14:59:22.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/2"}}},"commit":{"parents":[{"commit":"69c17319ccb40e1b209899ef4ba4b5a925983e3a","subject":"Fix Cinder integration and osbrick engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/69c17319ccb40e1b209899ef4ba4b5a925983e3a"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:59:22.000000000","tz":0},"subject":"Fix Nova integration","message":"Fix Nova integration\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973620\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973623\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/441c08a429f8aa245491c861dcf736aede8b30ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/441c08a429f8aa245491c861dcf736aede8b30ce"}]},"branch":"refs/heads/master","description":"Rebase"},"f2eec08c604ff06b9cb726549c995778d0ac0b14":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-16 15:01:23.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/3"}}},"commit":{"parents":[{"commit":"d00078db2336f9cc8d8151f9836e1853f134af37","subject":"Fix Cinder integration and osbrick engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d00078db2336f9cc8d8151f9836e1853f134af37"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 15:01:03.000000000","tz":60},"subject":"Fix Nova integration","message":"Fix Nova integration\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f2eec08c604ff06b9cb726549c995778d0ac0b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f2eec08c604ff06b9cb726549c995778d0ac0b14"}]},"branch":"refs/heads/master"},"68c320e07daadc752f5c8f9a53a8a4ee44ff1863":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-16 16:34:02.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/4"}}},"commit":{"parents":[{"commit":"d0bbaa8f07f50c1735eabb35bc12269231a00215","subject":"Fix Cinder integration and osbrick engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d0bbaa8f07f50c1735eabb35bc12269231a00215"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 16:34:02.000000000","tz":0},"subject":"Fix Nova integration","message":"Fix Nova integration\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/68c320e07daadc752f5c8f9a53a8a4ee44ff1863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/68c320e07daadc752f5c8f9a53a8a4ee44ff1863"}]},"branch":"refs/heads/master","description":"Rebase"},"6f018c9010bd0c7e2af821aa1eb4e1c99f256f17":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-16 16:39:11.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/5"}}},"commit":{"parents":[{"commit":"d0bbaa8f07f50c1735eabb35bc12269231a00215","subject":"Fix Cinder integration and osbrick engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d0bbaa8f07f50c1735eabb35bc12269231a00215"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 16:38:38.000000000","tz":60},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6f018c9010bd0c7e2af821aa1eb4e1c99f256f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6f018c9010bd0c7e2af821aa1eb4e1c99f256f17"}]},"branch":"refs/heads/master"},"7d2647b2ae46ba0b0b8a9ce94cbf04b4711d9ed1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-16 17:16:51.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/6"}}},"commit":{"parents":[{"commit":"697c28cfe06be90547dfccfc890f7bc059b28eab","subject":"Replace cinderclient with SDK and fix osbrick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/697c28cfe06be90547dfccfc890f7bc059b28eab"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:16:51.000000000","tz":0},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7d2647b2ae46ba0b0b8a9ce94cbf04b4711d9ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7d2647b2ae46ba0b0b8a9ce94cbf04b4711d9ed1"}]},"branch":"refs/heads/master","description":"Rebase"},"e9bb5b5d771788c4f47a4aa7bbb6b94c847a04ff":{"kind":"REWORK","_number":7,"created":"2026-01-16 17:22:22.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/7"}}},"commit":{"parents":[{"commit":"697c28cfe06be90547dfccfc890f7bc059b28eab","subject":"Replace cinderclient with SDK and fix osbrick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/697c28cfe06be90547dfccfc890f7bc059b28eab"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:22:12.000000000","tz":60},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e9bb5b5d771788c4f47a4aa7bbb6b94c847a04ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e9bb5b5d771788c4f47a4aa7bbb6b94c847a04ff"}]},"branch":"refs/heads/master"},"1c27be679eec3b6c56f1ccba5c0746379526fc5d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-19 08:36:14.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/8"}}},"commit":{"parents":[{"commit":"fc8c56e2b3a111862c95b9f958a10a22b2e9e1d4","subject":"Replace cinderclient with SDK and fix osbrick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fc8c56e2b3a111862c95b9f958a10a22b2e9e1d4"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-19 08:36:14.000000000","tz":0},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1c27be679eec3b6c56f1ccba5c0746379526fc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1c27be679eec3b6c56f1ccba5c0746379526fc5d"}]},"branch":"refs/heads/master","description":"Rebase"},"a7a09567e913e56c2e2ce9d674b81c569ee7b853":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-19 09:50:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/9"}}},"commit":{"parents":[{"commit":"d1d316ec6e6fd9d17c165c2e24fe6119b94d2b94","subject":"Replace cinderclient with SDK and fix osbrick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d1d316ec6e6fd9d17c165c2e24fe6119b94d2b94"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-19 09:50:41.000000000","tz":0},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a7a09567e913e56c2e2ce9d674b81c569ee7b853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a7a09567e913e56c2e2ce9d674b81c569ee7b853"}]},"branch":"refs/heads/master","description":"Rebase"},"50de97d3c90a84d9f63232dfe2e8061fe1030755":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-01-20 16:33:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/10"}}},"commit":{"parents":[{"commit":"28341fb23b2dccad8a112466724e1d20a1ae0c4d","subject":"Replace cinderclient with SDK and fix osbrick","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/28341fb23b2dccad8a112466724e1d20a1ae0c4d"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-20 16:33:26.000000000","tz":60},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/50de97d3c90a84d9f63232dfe2e8061fe1030755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/50de97d3c90a84d9f63232dfe2e8061fe1030755"}]},"branch":"refs/heads/master"},"6305f9574f7e9f68c0db44b5a7102e7d86eeb4e3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-01-22 14:04:29.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/25/973625/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/11"}}},"commit":{"parents":[{"commit":"5ee2923f22b71f0628b5f068910c0e8a685f1039","subject":"Replace cinderclient with SDK, fix cinder engines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5ee2923f22b71f0628b5f068910c0e8a685f1039"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-22 14:02:56.000000000","tz":60},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6305f9574f7e9f68c0db44b5a7102e7d86eeb4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6305f9574f7e9f68c0db44b5a7102e7d86eeb4e3"}]},"branch":"refs/heads/master"},"d38e46be9440d93a98e01ab6a1f7155617e39c76":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-01-26 16:58:40.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/25/973625/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/25/973625/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/25/973625/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/25/973625/12"}}},"commit":{"parents":[{"commit":"b56b3c4aa505d20397e682fc02d0fe018dd87bbf","subject":"Merge \"Fix Glance integration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b56b3c4aa505d20397e682fc02d0fe018dd87bbf"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:09:33.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-26 16:58:40.000000000","tz":0},"subject":"Fix Nova integration and change clients to SDK","message":"Fix Nova integration and change clients to SDK\n\nChange-Id: I9b2e1017d80695e1aea648e172c8def4dcff490d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d38e46be9440d93a98e01ab6a1f7155617e39c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d38e46be9440d93a98e01ab6a1f7155617e39c76"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
