)]}'
{"id":"openstack%2Ffreezer~973627","triplet_id":"openstack%2Ffreezer~master~I637a1c366492763f2af4b3b8ace0e107e5ba5a8d","project":"openstack/freezer","branch":"master","topic":"fix/nova-cinder-glance-integration","attention_set":{},"removed_from_attention_set":{"36848":{"account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"last_update":"2026-01-26 20:49:53.000000000","reason":"Change was submitted"},"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-01-26 20:49:53.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-26 18:33:26.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I637a1c366492763f2af4b3b8ace0e107e5ba5a8d","subject":"Replace cinderclient with SDK, fix cinder engines","status":"MERGED","created":"2026-01-16 14:18:41.000000000","updated":"2026-01-26 20:50:44.000000000","submitted":"2026-01-26 20:49:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"973627-fix/nova-cinder-glance-integration","meta_rev_id":"79b10f3a55c1f08d63d82fb3c7171f1fa4705543","_number":973627,"virtual_id_number":973627,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 20:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-01-26 18:33:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-01-26 20:07:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 14:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 19:50:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-19 19:50:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"state":"REVIEWER"},{"updated":"2026-01-19 19:50:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"8743728dd96df438e5a3049779118a0afcfeb61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14a47af18a830fcc28ef14fe3f686c4f981bb9a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:18:51.000000000","message":"Topic fix/cinder-integration removed","accounts_in_message":[],"_revision_number":1},{"id":"ac12d79fcfcc5e830e06da1b532d703de71ee800","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:18:57.000000000","message":"Topic set to fix/nova-cinder-glance-integration","accounts_in_message":[],"_revision_number":1},{"id":"44d74827fd7813b82c235de85281071ad3f68a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:49:00.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/c/openstack/freezer/+/973620, https://review.opendev.org/c/openstack/freezer/+/973623 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"57a52f5077e2b67d5f5396f23bcf257aca98798a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:59:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bf3693e607673eb0a8060e7b1da67f4eff86b2a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 15:01:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"796878371a46ade8e3ce1a390be58d7ece7f78f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 15:36:34.000000000","message":"Patch Set 3: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f62110f08eb481386c8235e3a978eaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24af06456734d189ed9128f20715fe0 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02011305f68e4f768c31062fe4124978 : FAILURE in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e8167ac1cfd46b2a2dbf23d7a9fbf50 : FAILURE in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66cdb5a5a8154669b71b2a125a96d246 : FAILURE in 6m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bab3bc60d90b41f7b77afbb6fdb3c02d : SUCCESS in 3m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/c8e2de748ea34aaeb6b6b6c4e9bf9e96 : SUCCESS in 16m 02s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e4b657bbb8204d558325dc6a733f0af0 : SUCCESS in 15m 45s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a094949dbe0044708d2d33507c88ba45 : SUCCESS in 19m 20s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/4b7509ac97a14fac844711a1d21849f6 : SUCCESS in 31m 44s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c5d44fadac54f73811429227274794d : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":3},{"id":"fc86c3d74ad6eafe7aeaceb1f9a8ca95fd15b224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 16:33:53.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2c6b7e96378f5d472614d9959f7710916c170aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 16:37:49.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1d8e0e9277ca2686b603be0ce3f8217c16c2a6fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:08:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0f9a317343f54c3389cd087b9c441c599629ba67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:13:30.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"77d0202c97ed84df92da89e92da3eb90358efbf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:16:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e72b4bae0d3a412861281e85674c1ee67e99eb7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:56:45.000000000","message":"Patch Set 8: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5281151d3f64b56b6b2da7e5e97d44f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7967f737b6ca42bbb9c8628d78f9d0ca : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712b28a4d1ef40768644b80378b949cf : FAILURE in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09b0cf2dc6084a8fa9e3e4fc7b0c5257 : SUCCESS in 7m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fe25310fef5470696bec3023b6d0590 : RETRY_LIMIT in 1m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77d60150362d48e4926cc138374eabb7 : SUCCESS in 4m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/dab1b9b1abb6497db8e02a569139c8e5 : SUCCESS in 36m 40s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/124bc42a844f4e0fb0cb899e262e6745 : SUCCESS in 17m 35s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a396aa47e50d4a1ba0e037c7a5c80128 : SUCCESS in 17m 34s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/f8b9a1f7e9ac4f559bedb6f8fd49203a : FAILURE in 25m 45s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1e91f7add234aa094493e33e31cda34 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":8},{"id":"65a77dfd56cde1e6618564f5f233507bef2ef629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 08:36:04.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"bc3657106fe8503d7c814bf6a7a233d0cd7a98e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 09:27:45.000000000","message":"Patch Set 9: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cc04e847bbe4131a5435fafa6f727e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7b036da556476585f8327d4ff24688 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ecd0a0552294458b21f907fc394b9d3 : FAILURE in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d747d3bb89674197882ca6ab88c717bb : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cd2c45e022b40998a2189a429075131 : RETRY_LIMIT in 3m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59496f21806b409aa9f1ded95ee1edfd : SUCCESS in 4m 08s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/58c9d37cab1c45b3921f2c4566e2e8be : SUCCESS in 19m 41s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b4545eed08ed4cd8bcb1afa98b31ec11 : SUCCESS in 18m 58s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af9d783d21154b4bb93a553033379206 : SUCCESS in 16m 22s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/248d51ec86c0436493cba8e80611de54 : SUCCESS in 41m 29s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebf606c0921a464c88c66fa904b04711 : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":9},{"id":"29170cb7f77cd7e33ee900ff7a638b72eca7b4fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 09:29:21.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4df89f1f6f78e2060e15b7eb43f3bdda64f952d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:07:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d79ad39b7d634646b4e8b16fa1d481d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da3fa28d5824379918f0c57765a7a46 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9351abdc42794d9aa5b83e952d83e193 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7a50daa5f89474d9ee9f170dffb74f3 : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2eefca2efcd456192d808a685f7c0f3 : RETRY_LIMIT in 1m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/16222981538d42dd848c003a0ea8f39b : SUCCESS in 2m 43s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9c9571d5643c420b9e51e1270f8c1874 : SUCCESS in 15m 50s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/15d8cf999c2e4d99bcc33566f108bebc : SUCCESS in 20m 36s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a22a65d675d485d885bc90cf70f08dc : SUCCESS in 32m 54s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/de05853ff82e44b480083682992c6ffc : SUCCESS in 23m 11s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1ff4fea13094a5db21f8aece042e732 : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":10},{"id":"612f6ddac3dd6af93c9071b0685d6b6a9b5a1f79","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-19 12:51:12.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a58903009b4bf0432d0d9ddd6fe795e2b7c190c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:36:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50e5c973355b4791b3fa8b7fc04f7a86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d61800320b14aec885a1134a4593b5b : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbeec25bd7574897babe4a31a0cfc3f6 : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2acbc076ac34f60afa6490e88518231 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/563ee8dfb9584201b8ebee60c0c29bbf : SUCCESS in 2m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/422bcb152a9941ef9e901ccee911695a : SUCCESS in 4m 40s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/cc8fc55bca184305a811ac4f449cfd7d : SUCCESS in 33m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/6e356dc098dd41e59e8425cf0efec7dc : SUCCESS in 18m 26s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a725feef3ef641f08807efbca975a255 : SUCCESS in 34m 00s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b3973b196bfa4df2a83eab27a6031470 : SUCCESS in 37m 46s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc082776a70745fd8b1fcd9ffee3c802 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":10},{"id":"9c2ba06e20ebe32c0a2ac2e3f75163a69046849b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-20 16:33:41.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"1559079ebf1e10d79149360af019a449b7b0f2de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:22:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461721f835e34a9c8ebc8e85895699dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e3412a093e4f9787a5f6507df6deef : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e872ccac9fb24c9f8fccc590808c8869 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80a88eb854d440318954dd516a8bf7b7 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9dba1c0642c44f6ba20871703f8d5b53 : SUCCESS in 2m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8058aca0024b45c2abe20c83031872f9 : SUCCESS in 3m 52s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e07fc37ea2974d5881598ed7d7624bf0 : SUCCESS in 33m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/89e7bf66216949c3bda0639efd6b3a11 : SUCCESS in 27m 14s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/839d96ad016e472e9d7b6e2f450d4809 : SUCCESS in 17m 43s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/2b2b387f86ec4b0a87db87645b4c52e7 : SUCCESS in 37m 12s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc3afecbc11c4a86823b3440a45ca187 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":11},{"id":"6024f8cda15d99146efad318495cbb201ce7de73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-22 14:04:29.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"e7c7fcb66cb4f3f4dcee9fb347192286b29438a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:09:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bf10d9fca3244e89f58740bdfc3f57c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2aa5ceee0f40ed8f88231b72784693 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010d7e97b2624eb697a6c97ab75f308f : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4ec6edc85334551a1837bd66bb7e527 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03e06d85f3f540f59a7a35497e03f750 : SUCCESS in 5m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93bb84ac4eb04bee8e43b2c48de3ae35 : SUCCESS in 2m 50s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2440493502884913a9f835076b90fb80 : SUCCESS in 37m 08s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/6ee3b223334b432180e60dac09798e66 : SUCCESS in 36m 09s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1a390962bf4b3ab630a5e8642241e5 : FAILURE in 39m 07s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/38f5072790a64a79b21ab4132688ce31 : FAILURE in 36m 56s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/26484f88473f4020bda57712e2bcc3c8 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":12},{"id":"c9c69afd9c1376272acf4ea03651b3aa61d1245f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 08:09:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"859b04f6197d3ee6154015873e9e76c4492beb35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 08:55:36.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aad9f2896fb4a69859a6771f00937d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5463bf698b28450982af4a90be3a130d : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5650cc1f6334a84a60a64c003ed626a : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/996e16d4aa8b4a3b883f2cf8a204086e : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6521be74e3934032a0514fc0782893f5 : SUCCESS in 4m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/628b211eabb9402c9cf3e3e72321fdcc : SUCCESS in 4m 04s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/0eaf2ec74e8e48449893a0d98ac58ec0 : SUCCESS in 31m 35s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/86632200dea649dcb54ea321879921f6 : SUCCESS in 36m 21s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6520a8322d7486495200d264de2899e : FAILURE in 34m 23s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/20df40d4757540129546233aa49cf162 : FAILURE in 15m 36s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/03902f65a24142d795e4088ffc0a00f2 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":12},{"id":"75cb1aa9d56e61f772b7398636f5dad032a36100","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 14:10:38.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"741c8a5d90ed0adc135b4a71f021c2b7001577d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:57:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f632dee250a4eb4a87fd0634b60656d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696df7500410466cb3c242306773ffb1 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae52eeecc5e474ab480ecba4f4e71c3 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b08534d637374647922e2f34b69128f7 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbd4403f13394709af85fb801fd85c6e : SUCCESS in 4m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/191927bc4a064be2bae0638f2c8b6959 : SUCCESS in 3m 51s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/53e1e218ba16470b9cf604b7f0de2bcc : SUCCESS in 34m 55s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/db4650e0112d4c5ea8046233874278cc : SUCCESS in 37m 23s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d94d4b90b29b4010be5ffbcb4e642ce6 : SUCCESS in 37m 16s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/dd129df28e33483492eeef8fab5d662f : SUCCESS in 21m 09s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/01b79796929b4baf97e939ee3f115e72 : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":12},{"id":"1fb85192ba15ef5faf6009c62bc4da07478ef2b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 18:33:26.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ffd034743b1cf01a2d389bf7d006bf92e349980d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 20:07:46.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"f40581f4356a87c1ea2c4c30f3c7c8e38340ef2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 20:08:17.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"625f3641ca912bb9c4fab4605f8a2f15591399aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 20:49:52.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f9ccc027efe40679f544cec26597356\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c895a839fe7b487f8c7eaba9a79b0873 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19d430840ce45e4b22914dbb5bc0d8a : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/209fa217814c4ebdb61b96b11a5263fe : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b7d841842ce471f9e641d9019051d46 : SUCCESS in 3m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/864523de3d3e4bcda825e7a30f19a819 : SUCCESS in 3m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d2fd23dc6c4a439d87b450712a42d80d : SUCCESS in 25m 40s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/63f433ca32a2453493ea69f7d14376a1 : SUCCESS in 34m 01s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c7f13248a3c4fafb9cd6612b1e6cc98 : SUCCESS in 35m 08s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/4aa2979e740a4340a7e135b95e26c6a4 : SUCCESS in 36m 08s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/10185bde8cdf45cab22a5ec893b4374e : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":12},{"id":"534028a4ef1acf28238f6d0ad0845cad7682f989","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 20:49:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"79b10f3a55c1f08d63d82fb3c7171f1fa4705543","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 20:50:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a42d26b224448ba6f95c93f1af19ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c9e58cb0424126a2e6fdaa3418a352 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"5ee2923f22b71f0628b5f068910c0e8a685f1039","revisions":{"6977c15b7d4e5f5f9b818247baf0b67c1322252e":{"kind":"REWORK","_number":1,"created":"2026-01-16 14:18:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/1"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:18:39.000000000","tz":60},"subject":"Fix Cinder integration and osbrick engine","message":"Fix Cinder integration and osbrick engine\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973620\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973623\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6977c15b7d4e5f5f9b818247baf0b67c1322252e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6977c15b7d4e5f5f9b818247baf0b67c1322252e"}]},"branch":"refs/heads/master"},"69c17319ccb40e1b209899ef4ba4b5a925983e3a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-16 14:59:03.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/2"}}},"commit":{"parents":[{"commit":"3448d7303dcf9b3c150cb54ed4ed1a7a0d7f7277","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3448d7303dcf9b3c150cb54ed4ed1a7a0d7f7277"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:59:03.000000000","tz":0},"subject":"Fix Cinder integration and osbrick engine","message":"Fix Cinder integration and osbrick engine\n\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973620\nDepends-On: https://review.opendev.org/c/openstack/freezer/+/973623\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/69c17319ccb40e1b209899ef4ba4b5a925983e3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/69c17319ccb40e1b209899ef4ba4b5a925983e3a"}]},"branch":"refs/heads/master","description":"Rebase"},"d00078db2336f9cc8d8151f9836e1853f134af37":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-16 15:01:23.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/3"}}},"commit":{"parents":[{"commit":"f30fd8bdbbf5394e7b1afaee6ee6db26e267f975","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f30fd8bdbbf5394e7b1afaee6ee6db26e267f975"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 15:00:59.000000000","tz":60},"subject":"Fix Cinder integration and osbrick engine","message":"Fix Cinder integration and osbrick engine\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d00078db2336f9cc8d8151f9836e1853f134af37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d00078db2336f9cc8d8151f9836e1853f134af37"}]},"branch":"refs/heads/master"},"d0bbaa8f07f50c1735eabb35bc12269231a00215":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-16 16:33:53.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/4"}}},"commit":{"parents":[{"commit":"5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 16:33:53.000000000","tz":0},"subject":"Fix Cinder integration and osbrick engine","message":"Fix Cinder integration and osbrick engine\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d0bbaa8f07f50c1735eabb35bc12269231a00215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d0bbaa8f07f50c1735eabb35bc12269231a00215"}]},"branch":"refs/heads/master","description":"Rebase"},"91a52095c4fe467bc8dd2008dc027550d391561e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-16 16:37:49.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/5"}}},"commit":{"parents":[{"commit":"5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 16:34:33.000000000","tz":60},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/91a52095c4fe467bc8dd2008dc027550d391561e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/91a52095c4fe467bc8dd2008dc027550d391561e"}]},"branch":"refs/heads/master"},"d0e9dbb57fff9951e24f4fe7a9c0732cc049a0e4":{"kind":"REWORK","_number":6,"created":"2026-01-16 17:08:35.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/6"}}},"commit":{"parents":[{"commit":"5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5e6e4f6d1201c5d9bb5e3bb6685e88a6e7ce65aa"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:08:23.000000000","tz":60},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d0e9dbb57fff9951e24f4fe7a9c0732cc049a0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d0e9dbb57fff9951e24f4fe7a9c0732cc049a0e4"}]},"branch":"refs/heads/master"},"9dbfd19db358c954de2cfa1609528181da6d39bf":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-01-16 17:13:30.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/7"}}},"commit":{"parents":[{"commit":"3e1548caab593e1fddd3b29d656c4a09e2582d05","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3e1548caab593e1fddd3b29d656c4a09e2582d05"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:13:30.000000000","tz":0},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9dbfd19db358c954de2cfa1609528181da6d39bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9dbfd19db358c954de2cfa1609528181da6d39bf"}]},"branch":"refs/heads/master","description":"Rebase"},"697c28cfe06be90547dfccfc890f7bc059b28eab":{"kind":"REWORK","_number":8,"created":"2026-01-16 17:16:20.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/8"}}},"commit":{"parents":[{"commit":"3e1548caab593e1fddd3b29d656c4a09e2582d05","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3e1548caab593e1fddd3b29d656c4a09e2582d05"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:16:08.000000000","tz":60},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/697c28cfe06be90547dfccfc890f7bc059b28eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/697c28cfe06be90547dfccfc890f7bc059b28eab"}]},"branch":"refs/heads/master"},"fc8c56e2b3a111862c95b9f958a10a22b2e9e1d4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-19 08:36:04.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/9"}}},"commit":{"parents":[{"commit":"581f220a442dfcc2a12ffe458f64e27b87b8b3ef","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/581f220a442dfcc2a12ffe458f64e27b87b8b3ef"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-19 08:36:04.000000000","tz":0},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fc8c56e2b3a111862c95b9f958a10a22b2e9e1d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fc8c56e2b3a111862c95b9f958a10a22b2e9e1d4"}]},"branch":"refs/heads/master","description":"Rebase"},"d1d316ec6e6fd9d17c165c2e24fe6119b94d2b94":{"kind":"REWORK","_number":10,"created":"2026-01-19 09:29:21.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/10"}}},"commit":{"parents":[{"commit":"581f220a442dfcc2a12ffe458f64e27b87b8b3ef","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/581f220a442dfcc2a12ffe458f64e27b87b8b3ef"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-19 09:29:08.000000000","tz":60},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d1d316ec6e6fd9d17c165c2e24fe6119b94d2b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d1d316ec6e6fd9d17c165c2e24fe6119b94d2b94"}]},"branch":"refs/heads/master"},"28341fb23b2dccad8a112466724e1d20a1ae0c4d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-01-20 16:33:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/11"}}},"commit":{"parents":[{"commit":"981de1b6d435586e85220c544d689efd4b040eeb","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/981de1b6d435586e85220c544d689efd4b040eeb"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-20 16:33:26.000000000","tz":60},"subject":"Replace cinderclient with SDK and fix osbrick","message":"Replace cinderclient with SDK and fix osbrick\n\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/28341fb23b2dccad8a112466724e1d20a1ae0c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/28341fb23b2dccad8a112466724e1d20a1ae0c4d"}]},"branch":"refs/heads/master"},"5ee2923f22b71f0628b5f068910c0e8a685f1039":{"kind":"REWORK","_number":12,"created":"2026-01-22 14:04:29.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/27/973627/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/973627/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/973627/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/973627/12"}}},"commit":{"parents":[{"commit":"981de1b6d435586e85220c544d689efd4b040eeb","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/981de1b6d435586e85220c544d689efd4b040eeb"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:16:45.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-22 13:55:57.000000000","tz":60},"subject":"Replace cinderclient with SDK, fix cinder engines","message":"Replace cinderclient with SDK, fix cinder engines\n\nReplaces the usage of the python-cinderclient with the SDK.\nFixes the osbrick engine by updating its volume attachment procedure.\nAlso addresses a bug that made the cindernative mode create backups\nvia cinder-backup that fail to restore due to invalid container names.\n\nCloses-Bug: #2138876\nChange-Id: I637a1c366492763f2af4b3b8ace0e107e5ba5a8d\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5ee2923f22b71f0628b5f068910c0e8a685f1039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5ee2923f22b71f0628b5f068910c0e8a685f1039"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
