)]}'
{"id":"openstack%2Ffreezer~975437","triplet_id":"openstack%2Ffreezer~master~I3c54dff267c4b137b501df379c23d8773b792879","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-04 16:04:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I3c54dff267c4b137b501df379c23d8773b792879","subject":"Fix restore-from-date behavior","status":"MERGED","created":"2026-02-02 17:54:48.000000000","updated":"2026-02-04 16:49:23.000000000","submitted":"2026-02-04 16:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975437","meta_rev_id":"158a259783e0fafdd61b28b88990de4007b4da48","_number":975437,"virtual_id_number":975437,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 16:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"all":[{"value":1,"date":"2026-02-03 10:35:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"value":2,"date":"2026-02-04 16:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"value":1,"date":"2026-02-04 16:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 18:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 10:35:18.000000000","updated_by":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"reviewer":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"state":"REVIEWER"}],"messages":[{"id":"504f55ebaee6dbacd30f8b4513dda4932d2c009c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-02 17:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bd6341479c863f468bd7a2ba35f6357b41063f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:43:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb9a4d5670d40fdbf17b75c0036df81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85809ae0da44e5cba6a523ea6e94294 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b06f21704694acfa45aac472fde95ad : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/264df5488c264dbab1879aadae1ec03f : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e1528a7c35044f3b516be96277492e3 : SUCCESS in 4m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af99d1c5669d47acb7162489349491ef : SUCCESS in 3m 53s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/dfd383ba99f04ab695191e79850165b3 : SUCCESS in 25m 17s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/20125d222bb444efa0fbf09e527699aa : SUCCESS in 15m 08s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb74aae3c6504a058d1518132b317351 : SUCCESS in 33m 57s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/90744b0d44fd46328b4c3dbe168a3a5a : SUCCESS in 41m 06s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd310ddaffbe4135a89a51bab68ac1b0 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b77ec435c16b8fdcba44dd9cd82bd059895b37e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-02 21:59:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0eb379a697e5ee97985ee6d9c60595069c01c286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 22:43:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d527d8e2234a83b2b4f765a711c81b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028f57db95c84b39aff35c6e6a51e000 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e240d45db9142038546710b090a25fe : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf30049709d4213b4ad60b5806b620a : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff965e410f3a47bcb4e28ded521517c8 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c3ce89a9c184351b4d44dd450e76eb6 : SUCCESS in 3m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dbe21e7ae51d44269cf5c09a391424c7 : SUCCESS in 3m 54s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ff9c1e0c7ba84197a55d4e80360f0a04 : SUCCESS in 33m 08s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/82ad834f76744fd195a4a6b03842aa5e : SUCCESS in 35m 17s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0805f16f60c45f084ed211db4cab036 : SUCCESS in 35m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/a8fdaa88bd8345d9a4c887cd6ce4061c : SUCCESS in 17m 37s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/37db9081862d49b7865beece7ca20d62 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"9f467e41f1995928e240e27dfa1c694e5cb5a40e","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-02-03 10:35:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"665e01109fcb137dc4b54d990c91fbf3f9ddcf8d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-04 16:04:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c771240db564bb0db69140f0cc8e71988c0ce4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:05:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19e4d88edc8329ee387d353c84651f8a241aca32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:48:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c5f61e212af4cf8a4a2794267542cd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8a23cdb08f4b8da692da9781c5f227 : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71759ebc870849c3a81f4947d52d3eae : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2107d69db9144240b3a628d5669c1cd1 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/998f7379697f43e3be01b2872a3061d0 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d21bbe62b5f84633892fa2f48bb19654 : SUCCESS in 3m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb7e6b7b9dfd4323b48f2fdefeb6a870 : SUCCESS in 2m 32s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2ddf082de5b049ab9fad89a614f2cd6c : SUCCESS in 35m 08s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/5a3b7cd75ca94a5f8d6a55ee81bdd27d : SUCCESS in 33m 08s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d12fcc4c5726439ea61f94312e6eb017 : SUCCESS in 35m 33s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b0b16cd481704c82a27f2ec7639a2480 : SUCCESS in 17m 04s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bd9757bb34b470aaabe0d21f0911f99 : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e99e1973ad4bef8b5ade1f07b6544263bf6a3336","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:48:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"158a259783e0fafdd61b28b88990de4007b4da48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:49:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d18f618e7a64f9ab68e65f001b1798a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1034723ec554606b7192b3a2f55dceb : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e0a09e3d2064c3e864c9f592df619b9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"496413a74b891dd0bd25eccfc0ceb463349c8259","revisions":{"840c5cd8a1a327e161c2d8b6c1ea6f2d3977f47f":{"kind":"REWORK","_number":1,"created":"2026-02-02 17:54:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/975437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/37/975437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/37/975437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/37/975437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/37/975437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/37/975437/1"}}},"commit":{"parents":[{"commit":"a3210ca85aa698e0af957dd85b41609c6522261e","subject":"Merge \"Consistent \u0026 meaningful OpenStack resource naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a3210ca85aa698e0af957dd85b41609c6522261e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-02 17:54:41.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-02 17:54:41.000000000","tz":60},"subject":"Fix restore-from-date behavior","message":"Fix restore-from-date behavior\n\nAt the moment the logic of `--restore-from-date` is reversed from the\ndocumented one, which leads to unexpected behavior.\n\nWe restore the behavior of freezer according to the documentation\nalong with adding a unit-test which covers\nthe behavior to avoid regressions in the future.\n\nCloses-Bug: #2138883\nAssisted-By: Gemini 3\nChange-Id: I3c54dff267c4b137b501df379c23d8773b792879\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/840c5cd8a1a327e161c2d8b6c1ea6f2d3977f47f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/840c5cd8a1a327e161c2d8b6c1ea6f2d3977f47f"}]},"branch":"refs/heads/master"},"496413a74b891dd0bd25eccfc0ceb463349c8259":{"kind":"REWORK","_number":2,"created":"2026-02-02 21:59:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/975437/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/37/975437/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/37/975437/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/37/975437/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/37/975437/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/37/975437/2"}}},"commit":{"parents":[{"commit":"a3210ca85aa698e0af957dd85b41609c6522261e","subject":"Merge \"Consistent \u0026 meaningful OpenStack resource naming\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a3210ca85aa698e0af957dd85b41609c6522261e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-02 17:54:41.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-02-02 21:59:29.000000000","tz":60},"subject":"Fix restore-from-date behavior","message":"Fix restore-from-date behavior\n\nAt the moment the logic of `--restore-from-date` is reversed from the\ndocumented one, which leads to unexpected behavior.\n\nWe restore the behavior of freezer according to the documentation\nalong with adding a unit-test which covers\nthe behavior to avoid regressions in the future.\n\nCloses-Bug: #2138883\nAssisted-By: Gemini 3\nChange-Id: I3c54dff267c4b137b501df379c23d8773b792879\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/496413a74b891dd0bd25eccfc0ceb463349c8259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/496413a74b891dd0bd25eccfc0ceb463349c8259"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
