)]}'
{"id":"openstack%2Ffreezer~980438","triplet_id":"openstack%2Ffreezer~master~Iff6008680f23c11cc77e9629bea1414bd420d565","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-19 11:56:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Iff6008680f23c11cc77e9629bea1414bd420d565","subject":"Distribute freezer scheduler config sample","status":"MERGED","created":"2026-03-13 14:25:29.000000000","updated":"2026-03-19 12:38:16.000000000","submitted":"2026-03-19 12:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980438","meta_rev_id":"f4a0e3f4599e01b5681d58bf762ef4ef74167657","_number":980438,"virtual_id_number":980438,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 12:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-19 11:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-16 15:33:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-03-19 11:56:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 15:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-16 15:33:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e6aadbf514ec632ac5101445f88b4b6f5b735cba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 14:25:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c90f51ac66f3e97ce0e558a767817868f4f50c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 15:09:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d83b67f206499ba7f0fdde4238f4fc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd23e71bca6443d68bbe0aa590869be0 : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/145b681f19f34587b2c0bc7307e929c8 : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ed580ca98740b4b98768fd356f08bb : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4bd25d787d4b22b3d116805f52d12f : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c4013413d47449399786031de32e5e4 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5478690c70ad4aacbeddd31846eb792a : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d807ed7c3d641df9299880fefd701f3 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ca87b558b1949ffb4fa420c8aa206c2 : SUCCESS in 2m 23s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/1f2a539e6b8f456984517eeed68c15b7 : SUCCESS in 29m 42s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3abff56a8cbe4a2db0f82968819edfcb : SUCCESS in 20m 57s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2ba151197d9441fa646eb6a9e1861dc : SUCCESS in 24m 57s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9f5332cff89e42b5bb60fc13ea47370e : SUCCESS in 38m 40s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/01e32fc813f14690b1fb363afbf36ddd : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"439637518c34c3dac75ba4b305a87872ffa5129d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 15:33:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cc9f3c0fcb771fed2c9b968f8f196a077834bb88","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-19 11:56:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee91a002e14b9385a526b3d5159280552f27873b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:57:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"256ed0e3c1f9d1de65dd441fae865b418ed59f6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:37:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e7aea3097e04dbba6bbc83e5e30b41f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2204cfb9796a40f2a5f573edbb568597 : SUCCESS in 3m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b3d6ce44c9c44e983d6e76e9f1878ed : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7a6ed07ee249078c2a9e8225399c6c : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0ff9ea2ada4d94870d6419dac0f831 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c339b050d8f94a3eaf3c7127a6c52bf5 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a6deb3b0ea343e7848c92ec0b967596 : SUCCESS in 3m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/416616e6dad34afcbd7c7204d1c6b44e : SUCCESS in 4m 51s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2dde324af733488eb1aa7f8ab672a9b9 : SUCCESS in 14m 20s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/2d87aed87d9748738293b5e0c683e49f : SUCCESS in 34m 06s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b014b005ce3465bad8730447fd32ff5 : SUCCESS in 22m 04s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/3dbc6d32f7584243a15e92f834d556db : SUCCESS in 35m 46s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/c204a5f9c53f4a60b8082aa585c3fbb0 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b1758b3eafa417d81ecc14a4b2d418065ae7d8fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:37:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f4a0e3f4599e01b5681d58bf762ef4ef74167657","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:38:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a53fcfaea844318b9a5b9ebdc968ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685d1a9ec91f4036b3758d9987792abc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4706e54d9c95f1d75009ef444121fec621d46f6","revisions":{"b4706e54d9c95f1d75009ef444121fec621d46f6":{"kind":"REWORK","_number":1,"created":"2026-03-13 14:25:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/980438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/38/980438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/38/980438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/38/980438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/38/980438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/38/980438/1"}}},"commit":{"parents":[{"commit":"b9cad5a6b60c424250b205e62655a14a9a4d7180","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9cad5a6b60c424250b205e62655a14a9a4d7180"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 14:25:22.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 14:25:22.000000000","tz":60},"subject":"Distribute freezer scheduler config sample","message":"Distribute freezer scheduler config sample\n\nIn order to simplify a life of operators, let\u0027s install sample config\nfile for freezer scheduler, which provides configuration reference.\n\nChange-Id: Iff6008680f23c11cc77e9629bea1414bd420d565\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b4706e54d9c95f1d75009ef444121fec621d46f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b4706e54d9c95f1d75009ef444121fec621d46f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
