)]}'
{"id":"openstack%2Ffreezer~982082","triplet_id":"openstack%2Ffreezer~master~I385f46bac6d458b52290d430030c9226b90083a3","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I385f46bac6d458b52290d430030c9226b90083a3","subject":"Add a separate section for keystone auth","status":"MERGED","created":"2026-03-25 13:01:09.000000000","updated":"2026-03-26 11:42:40.000000000","submitted":"2026-03-26 11:41:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982082","meta_rev_id":"502091c05059e65a19a935dbcdf802fdbd80280c","_number":982082,"virtual_id_number":982082,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 11:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-26 10:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-03-26 10:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-25 13:45:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa43fda6a4552f905537a2e91466ad9c5b231430","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-25 13:01:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6920ffdea261f3621253ae88965eb97c52ebdbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:45:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82b55789a5284e3baf83414af9632c58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463bec573fb443fdaf9829edcaeef39d : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70378390b1b145d1b7477a0fac22a7fd : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1233a0bf4da24e34a1aab84706a316e6 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82e64ae8e4da45d0bbea79813b892cc7 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61c467895df14ced9c8852072fb16f43 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1103704418004dcea23dc4755f8d0ff3 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66c9df3dd9e747b3a1a190f7692ce1b7 : SUCCESS in 5m 46s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9ef963390b5e4b2bac21453a0bdbef2a : SUCCESS in 18m 19s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/40ece7186cef4edc9a94f370d3f31b60 : SUCCESS in 18m 15s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c1178386d2640a889e0d24a4462555b : SUCCESS in 36m 16s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/8d921a510d6c41e29fde825af105fe3b : SUCCESS in 38m 26s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5f82ef7f84747e38f0b2eaf7db5a8b7 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e5ea31a0badd33e9614a0371d18f5a08cf52167d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 10:57:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a585a401476477bcd26813394b36fe5baf9da40f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 10:57:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccda46eda1abe80951bffee6782ef5fc8560a338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:41:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9ed0c3fb864556a89a3f7cb97714b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3cd7090cc749a2b08013514b48a04a : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6115c38c6b1f4a599104227781a3d348 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03612b3865994edfb5061107152e7376 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c28fa95ceb68427cae63bb9d23d0550f : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/968ba1ad116e4d2da22b31ece26a5ba3 : SUCCESS in 3m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/470add2cd5f14a538b1c6e4deb726d81 : SUCCESS in 3m 15s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8fab49997b2345b8ab3bca8048c7ef5b : SUCCESS in 33m 59s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/6f045039bc9a4f56943cf7ef71844278 : SUCCESS in 38m 52s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dd96275dab44b4980f6833dc1ee1bfe : SUCCESS in 15m 07s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/74aa1e83da2f43e5b4b2e1e76c65631f : SUCCESS in 29m 13s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/39b8c9d4673f44d79c27e973ea19a30c : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ff000292539c2433a81a2612b64cac753b7fb0f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:41:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"502091c05059e65a19a935dbcdf802fdbd80280c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:42:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f88a3f08fda94f0abc2a95aee463f5d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea16a1ed80b4421bafdcf411ec38c479 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5c4aad40d6440ca9b6ecba2aa06aa07 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"536128e81c912e67e34de767d70be62dad511d3a","revisions":{"536128e81c912e67e34de767d70be62dad511d3a":{"kind":"REWORK","_number":1,"created":"2026-03-25 13:01:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/982082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/82/982082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/82/982082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/82/982082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/82/982082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/82/982082/1"}}},"commit":{"parents":[{"commit":"ca4c0a8c6c484805719fa6e28cab4c62f6235aa9","subject":"Merge \"Default to identity version 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ca4c0a8c6c484805719fa6e28cab4c62f6235aa9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-25 13:01:01.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-25 13:01:01.000000000","tz":60},"subject":"Add a separate section for keystone auth","message":"Add a separate section for keystone auth\n\nDefining keystone auth details in DEFAULT is non-conventional\nand may be confusing for end users.\n\nThus, let\u0027s follow best practices and add a separate section for\nkeystone details, while keeping backwards compatibility.\n\nChange-Id: I385f46bac6d458b52290d430030c9226b90083a3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/536128e81c912e67e34de767d70be62dad511d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/536128e81c912e67e34de767d70be62dad511d3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
