)]}'
{"id":"openstack%2Ffreezer~982215","triplet_id":"openstack%2Ffreezer~stable%2F2026.1~Ib2ecd6e7295cfd96f7786db7b605842160f068a3","project":"openstack/freezer","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-30 16:30:02.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ib2ecd6e7295cfd96f7786db7b605842160f068a3","subject":"Fix scheduler daemonization and file descriptor preservation","status":"MERGED","created":"2026-03-26 09:19:04.000000000","updated":"2026-03-30 17:54:19.000000000","submitted":"2026-03-30 17:52:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982215","meta_rev_id":"efb36baf38dd83f36b407f1a43ba18bd77a212fb","_number":982215,"virtual_id_number":982215,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 17:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-30 15:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-03-30 15:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-26 09:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c75a4b5927e1e8fea0cd66721890364f88d04601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 09:19:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6030f2a4f504b5a1e4a088795aae65701e813c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 09:43:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fc5f09c80a64b9496eb487494f870e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a2ad59223f4fae898face97964a384 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9efa253704e4193a3382ff4d4f4db7f : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/418d9d545b8943f286dd159a70fe7967 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0165b4905ef478b89ecb8b405f3b670 : SUCCESS in 4m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a2ce71af5dd40b0a71c9357bf6556ff : SUCCESS in 3m 43s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9ea6dc3c8017442795d6e37dd31d0efa : FAILURE in 12m 00s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/4a84f2fee91046c09d63715f61ac797c : FAILURE in 13m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3aa2b844d0e44b994ab6458efd29374 : FAILURE in 19m 38s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/1d48c2e0c3d244ffa3b018d85ac651da : FAILURE in 11m 41s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f49b8ca17ab9427db89dfbcbbb4e4e6c : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"74cd5b6f11b76f0436d0d54dbca47593a2449268","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 10:54:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f60058cf6ac0499c3e5109d9a408fbb68c5154a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:31:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c968842de19d402ab1f1f015cb88b047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7296b1b5fe624c48a6c36c58f7d3d68f : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8417c900575b49418ac87849699984a1 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/067bda3d3ee4413695c7b053f10ccf1c : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adbef19721c04970ab52a69c50da3b15 : SUCCESS in 4m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cc2043d22284256bcd6402e6132c70b : SUCCESS in 3m 43s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/699d54679adb44ab96593c54fc9cd274 : FAILURE in 19m 57s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/44d967d0f57645edae0cd284bdbaab2d : FAILURE in 21m 56s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e0d17eaef6342c58cd3bc8cc853a967 : FAILURE in 13m 16s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/a00d00dce6f4423387421ba950b5d8bd : FAILURE in 9m 08s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/11931ff703a84d38a9bbc12fb24e1a87 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b70f4b893ba716761dbf31610b35c62a9a79c22c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 13:30:10.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d60344e881a8e5c74038dd0ccee605d2fd576676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 14:22:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f07d91cf2440b2886916d7d4bf0630\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f58d0752305249f7a9dcf7c172dcb1b8 : SUCCESS in 2m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec44e594d50f42aa820e1d6098ee39c6 : SUCCESS in 1m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a840637b1fc34738802edef523a74e6a : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06afff00e31447a59e701e6734797ecc : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47221f4e51f4439e96700c8c63fef78b : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b88de42a3344755be358889faa7a0d5 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64ee42967d4946fc95796648250d4134 : SUCCESS in 3m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c86c91864eff4ce1b611ce740ee8a11f : SUCCESS in 3m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/fa40184e50bd43ebbf61ed872b54bd6d : SUCCESS in 22m 52s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/014f2cd1806d439ebd8171b56df3ec7c : SUCCESS in 36m 15s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d5a8fd6a34241a3b7620a10cd89df10 : SUCCESS in 19m 00s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/63f8b8e44b824610beb8d33d77163490 : SUCCESS in 22m 27s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/18710b10c7a848fa8e627fbc2eb64235 : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"96b22869f223dbfb716fcac15405d8c082b985aa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 15:18:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a8fd13de14678652a51bef8be775ef111dc9bcd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:19:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe4a2f296d82ba5ab45cd76884a74511dddb50e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:03:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c42edaacc7d4006ab37f87ab2580419\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c3481703f6d43d4802ce7621d2ecd6e : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a696a466861403da9ca4c881510c7bc : SUCCESS in 1m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76dab4ebb7549a18c6cc1a0b7aa6c49 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a4b9849ac064792bed24d95969beff3 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318f58ce85e4418f9bdc99be9611d317 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8504be14007464ca458fe07e76fd488 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5373f7de98f40ebb70f1973dbe1685f : SUCCESS in 3m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fdb39b6131c44c55bfb81bbdc2cc257f : SUCCESS in 3m 57s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e59cbfb0c6d446d5aefaf53a8326748b : FAILURE in 33m 50s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1a90d029b63e4ad9b673e6df56d8cb28 : SUCCESS in 34m 58s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/941f34f9f4ab47fc9bae46c5bd28f849 : SUCCESS in 38m 33s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/e3a240b258654a3198979871912800ab : SUCCESS in 36m 07s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/c60d1d6709d54b7eb9e32aadc58e4d37 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"62aaf021cf4945157c484ac42c0f9f96fadb6065","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 16:30:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7f89d9990c309acd0e587f7a773a27e0cd6dc0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:10:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0c30246836641b4978c02de3c1eec5f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d04cc48b372641b58d28abf629e259ec : SUCCESS in 2m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/79686948c8fa4c379e57ffebe3a1a0de : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b22f5a63a7471c81124e19e684f988 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8baeb72ab1649d1b677569a0fedd570 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/073c4855c55f470ebbd2df6bf4e69ce0 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07db5257c42844bb89a4a507cfbc8899 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26ddcd6c633141239cc8f444f9877f26 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf15ebb166d64ff98c101f79bdb6f30d : SUCCESS in 3m 11s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8f61920737934e8b88025a932b087b6c : SUCCESS in 35m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e2172db344214c6ba3beb750d053a6c9 : SUCCESS in 35m 37s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16f33f2d3c9c404ba8a0a726e0b3d374 : SUCCESS in 37m 32s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9c418e4a6508477085ba26b421a12182 : SUCCESS in 19m 58s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaac3f411cf14038ba0665cc94ed1e14 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":2},{"id":"54f67e15d5ef228ffe53e590135b0153e1871861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:11:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e7c16e2d40f3851336196d5c8b9b283766582b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:52:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c75ee52af948108093a65d151b8ccd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6174975aab34fa2b26a306ffcbfaff2 : SUCCESS in 3m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3bcb670611ee4bb38cb3a1fc42e278bc : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030f8849ef924a4d804ca8a807462a4b : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76403e89443c47cf9d6e0b6568d53c05 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7da12f40d9407785e69fca84c694d0 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/863a5c94296a4a14b8dce1087abbfc03 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eacd3c76b505469a87756056a8f04101 : SUCCESS in 2m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b41da17d98e54b28a0d6b790e4cdf9a6 : SUCCESS in 5m 26s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/53e13fad03f84ae78873574d4e83041e : SUCCESS in 31m 18s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/60b13c0c1db54dd7bddd235c51a3ddb2 : SUCCESS in 29m 22s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74bad9b5a708463084276cfd3eb85da8 : SUCCESS in 35m 59s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9167f48ecc78480ba4bcdf8b41fbc984 : SUCCESS in 35m 28s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2ea7281fcf84f47af88e5c8d04f72bd : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"11e91059dc49f755ea2b8de46a06c4e7f7034f35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:52:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"efb36baf38dd83f36b407f1a43ba18bd77a212fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:54:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c16b485acd4f3391c059fd64316994\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8401e71d15b54f81a74d3aeacf557d35 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ecc40806668188e979c286242892608e0bdc4af5","revisions":{"d23d7c7964c11f43745d506ae3a2ff1aa377e9d7":{"kind":"REWORK","_number":1,"created":"2026-03-26 09:19:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/982215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/982215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/982215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/982215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/982215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/982215/1"}}},"commit":{"parents":[{"commit":"ea090d8c75e24bcf0e0e997bccdcb4a9bd45c977","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ea090d8c75e24bcf0e0e997bccdcb4a9bd45c977"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-25 19:37:05.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-26 09:18:57.000000000","tz":60},"subject":"Fix scheduler daemonization and file descriptor preservation","message":"Fix scheduler daemonization and file descriptor preservation\n\nFreezer Scheduler launched in daemon mode was silently failing\nwhen configured to use journald as logger.\n\nThis squashes changes in [1] which were\ninsufficient as it introduced several issues:\n\nIt attempted to re-initialize logging inside the DaemonContext\nusing a nested import and setup_logging() call.\nThis could lead to deadlocks or unexpected behavior when\nfile descriptors were being aggressively\npreserved/closed by the daemon library.\n\nThis is fixed now with a wider refactoring of the daemon mode logic\nand how work with sockets is being performed.\n\n[1] https://review.opendev.org/c/openstack/freezer/+/982132\n\nAssisted-By: Gemini 3.1 Pro\nChange-Id: Ib2ecd6e7295cfd96f7786db7b605842160f068a3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d23d7c7964c11f43745d506ae3a2ff1aa377e9d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d23d7c7964c11f43745d506ae3a2ff1aa377e9d7"}]},"branch":"refs/heads/stable/2026.1"},"ecc40806668188e979c286242892608e0bdc4af5":{"kind":"REWORK","_number":2,"created":"2026-03-30 13:30:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/982215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/982215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/982215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/982215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/982215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/982215/2"}}},"commit":{"parents":[{"commit":"ea090d8c75e24bcf0e0e997bccdcb4a9bd45c977","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ea090d8c75e24bcf0e0e997bccdcb4a9bd45c977"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-25 19:37:05.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 13:30:10.000000000","tz":0},"subject":"Fix scheduler daemonization and file descriptor preservation","message":"Fix scheduler daemonization and file descriptor preservation\n\nFreezer Scheduler launched in daemon mode was silently failing\nwhen configured to use journald as logger.\n\nThis squashes changes in [1] which were\ninsufficient as it introduced several issues:\n\nIt attempted to re-initialize logging inside the DaemonContext\nusing a nested import and setup_logging() call.\nThis could lead to deadlocks or unexpected behavior when\nfile descriptors were being aggressively\npreserved/closed by the daemon library.\n\nThis is fixed now with a wider refactoring of the daemon mode logic\nand how work with sockets is being performed.\n\n[1] https://review.opendev.org/c/openstack/freezer/+/982132\n\nAssisted-By: Gemini 3.1 Pro\nChange-Id: Ib2ecd6e7295cfd96f7786db7b605842160f068a3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ecc40806668188e979c286242892608e0bdc4af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ecc40806668188e979c286242892608e0bdc4af5"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
