)]}'
{"id":"openstack%2Ffreezer~982427","triplet_id":"openstack%2Ffreezer~master~Ib13254f32236a348f575d217253874053acb7b16","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-30 15:20:14.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ib13254f32236a348f575d217253874053acb7b16","subject":"[doc] Auto-generate CLI options","status":"MERGED","created":"2026-03-27 14:19:50.000000000","updated":"2026-03-30 16:23:25.000000000","submitted":"2026-03-30 16:21:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982427","meta_rev_id":"ed7aec724ca8311e07bb234d74a0c5c61a6199ae","_number":982427,"virtual_id_number":982427,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 16:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-30 15:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-03-30 15:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 15:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d0232a5e96770e7ab358690521cf2b7bdf98d21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 14:19:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60eda6188cdcf289399e8442cbae4ee5f6ea4c04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 14:24:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"692aecfaa52708f9ccc78defae860e43c15d373a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 14:24:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"354a5f84e8262a4bee3c1cda5a25bca02b6f63a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-27 14:36:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"33f4602288e95c03db9e4fc3ce20f53a02b4199e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 15:25:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36a6a71f932246309af385663f29a6bf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/887c3dd0621442ef9e1b74a8ae58563e : FAILURE in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f259b64b71747178811ca505fc2ef12 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32e1f6258d894fb2a07688d4ddf012e7 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e7c33f38e1b40dfb73dce4511e72c34 : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/744f525fbc6f4a98858b6152b3bf1852 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8656df9ceac404e9304e7e615353638 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa98032c7b5f45e3b34658122d9b7a06 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3d3d486384c4738b447ab84f2a2bbd5 : SUCCESS in 3m 53s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/450bc2fa702047d7a728997ad7dd19d8 : SUCCESS in 15m 21s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/dfc2658290fd49f8a50cd0b132aebc39 : SUCCESS in 43m 24s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9fb404f103c47b2b431e021519a099d : SUCCESS in 35m 04s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/69355caf257340e69ce06ed26d700db3 : SUCCESS in 25m 54s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c142529f3fd49329aeb5ef5375080b4 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"5f7f3f2d2bb3765bfd6b8a37b7ade66d7b4236cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 13:47:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fdcaa8c576b4c03a05e3e556886a8eac5841535f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 14:41:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f41701d33984fcb9eb5ea176c5fb28a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/76026b0b452c44719e57007be5268cfa : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3c2f53d3c94a33a0a38034880dea4b : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c099324cb4de4a34b76bbdbd4d558036 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853593493ea24358b79e920257826bc2 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7a40f4ee526492da7702e73ea763763 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5d5f44ee2f543b78e665dfcb62e2a94 : SUCCESS in 3m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ecd3a8a1e4544e9c8ff972f40e1878f1 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ecc3c8e1d194bd9bdb33a5bf9806e90 : SUCCESS in 4m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e8327f9014aa4f0fb72bca343a492526 : SUCCESS in 15m 51s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/fb4a0bc22f004af3a371c6ecc8dc017d : SUCCESS in 18m 55s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25392d4de02145d59af1ba46dd3d49e7 : SUCCESS in 46m 15s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b7f8eb30df2942e8825849d7ae045f5f : SUCCESS in 21m 38s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/cab611cd2f8e428a8b8bb05b5aa9346a : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":5},{"id":"285614c0d52c2a263506c185ee8c184560052345","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-30 15:20:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"79d12d801c0c93b71eb5939d718dab5f2b1605c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:20:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"87b3e8022ff8078c99a8a30f9a89a95d594be182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:21:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b124d95517476ca95397f579788b4f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/84b26264864b4ff3a9ce96a5ce5e180a : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b4b85ad35241d3a1eabe104e02374b : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/884cc561c2ca495db4b07d1d6f7341b8 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8fb321206864ff4b4395d87bac48af6 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0de44f523b5947f7aa7269ca469aeaa5 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79984544381c459b86bb9ad080e1c458 : SUCCESS in 3m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cfb52f0e2d754cc6b3c4c24da06d2f67 : SUCCESS in 4m 02s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/4c45b791bfbf4005ae93a785640fa731 : SUCCESS in 18m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ebe9aa273aa6495b82f2a2a4a72db976 : SUCCESS in 15m 35s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90a59a3b85544064ab4c20d32c72df27 : SUCCESS in 30m 38s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/ebccfe161eb14b65a40d18aa884e82da : SUCCESS in 27m 48s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/6167f954e1824773a267634d88b3c383 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f4b225cf027d6d341068b5892dc11d1f04a0a7c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:21:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ed7aec724ca8311e07bb234d74a0c5c61a6199ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:23:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc604560a3c8461195c05d12eda90e3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3ebc9ea1fb45e49eec7b4371d0293d : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/248e4e5e4bcd4d329bcddaf5b6f6be10 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"90f7e706cecdb1f5a8031a22ee72cc950148fd93","revisions":{"decf63ba00e7ffa9cb16cfba6cdfeb17a2a85358":{"kind":"REWORK","_number":1,"created":"2026-03-27 14:19:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/982427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/982427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/982427/1"}}},"commit":{"parents":[{"commit":"3aea592be1a4756d3c35992df26119fc891560c2","subject":"Move scheduler options to own group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3aea592be1a4756d3c35992df26119fc891560c2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:19:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:19:44.000000000","tz":60},"subject":"[doc] Auto-generate CLI options","message":"[doc] Auto-generate CLI options\n\nInstead of manually updating documentation, let\u0027s leverage autoprogram\nextension for sphinx, along with a simple CLI parsing script, to\nCLI arguments for tools in automated manner.\n\nChange-Id: Ib13254f32236a348f575d217253874053acb7b16\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/decf63ba00e7ffa9cb16cfba6cdfeb17a2a85358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/decf63ba00e7ffa9cb16cfba6cdfeb17a2a85358"}]},"branch":"refs/heads/master"},"490af8f495b8680f6456d87d4f670d5788280781":{"kind":"REWORK","_number":2,"created":"2026-03-27 14:24:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/982427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/982427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/982427/2"}}},"commit":{"parents":[{"commit":"3aea592be1a4756d3c35992df26119fc891560c2","subject":"Move scheduler options to own group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3aea592be1a4756d3c35992df26119fc891560c2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:19:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:24:13.000000000","tz":60},"subject":"[doc] Auto-generate CLI options","message":"[doc] Auto-generate CLI options\n\nInstead of manually updating documentation, let\u0027s leverage autoprogram\nextension for sphinx, along with a simple CLI parsing script, to\nCLI arguments for tools in automated manner.\n\nChange-Id: Ib13254f32236a348f575d217253874053acb7b16\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/490af8f495b8680f6456d87d4f670d5788280781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/490af8f495b8680f6456d87d4f670d5788280781"}]},"branch":"refs/heads/master"},"b9d9919876bb1db9a024dd0100cdf5174a2d7834":{"kind":"REWORK","_number":3,"created":"2026-03-27 14:24:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/982427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/982427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/982427/3"}}},"commit":{"parents":[{"commit":"3aea592be1a4756d3c35992df26119fc891560c2","subject":"Move scheduler options to own group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3aea592be1a4756d3c35992df26119fc891560c2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:19:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:24:41.000000000","tz":60},"subject":"[doc] Auto-generate CLI options","message":"[doc] Auto-generate CLI options\n\nInstead of manually updating documentation, let\u0027s leverage autoprogram\nextension for sphinx, along with a simple CLI parsing script, to\nCLI arguments for tools in automated manner.\n\nChange-Id: Ib13254f32236a348f575d217253874053acb7b16\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9d9919876bb1db9a024dd0100cdf5174a2d7834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9d9919876bb1db9a024dd0100cdf5174a2d7834"}]},"branch":"refs/heads/master"},"3df64fc61e48ff416c4c26329e04a58a84e030c3":{"kind":"REWORK","_number":4,"created":"2026-03-27 14:36:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/982427/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/982427/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/982427/4"}}},"commit":{"parents":[{"commit":"3aea592be1a4756d3c35992df26119fc891560c2","subject":"Move scheduler options to own group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3aea592be1a4756d3c35992df26119fc891560c2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:19:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:36:02.000000000","tz":0},"subject":"[doc] Auto-generate CLI options","message":"[doc] Auto-generate CLI options\n\nInstead of manually updating documentation, let\u0027s leverage autoprogram\nextension for sphinx, along with a simple CLI parsing script, to\nCLI arguments for tools in automated manner.\n\nChange-Id: Ib13254f32236a348f575d217253874053acb7b16\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3df64fc61e48ff416c4c26329e04a58a84e030c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3df64fc61e48ff416c4c26329e04a58a84e030c3"}]},"branch":"refs/heads/master"},"90f7e706cecdb1f5a8031a22ee72cc950148fd93":{"kind":"REWORK","_number":5,"created":"2026-03-30 13:47:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/982427/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/27/982427/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/27/982427/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/27/982427/5"}}},"commit":{"parents":[{"commit":"3aea592be1a4756d3c35992df26119fc891560c2","subject":"Move scheduler options to own group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3aea592be1a4756d3c35992df26119fc891560c2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-27 14:19:44.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-30 13:47:04.000000000","tz":120},"subject":"[doc] Auto-generate CLI options","message":"[doc] Auto-generate CLI options\n\nInstead of manually updating documentation, let\u0027s leverage autoprogram\nextension for sphinx, along with a simple CLI parsing script, to\nCLI arguments for tools in automated manner.\n\nChange-Id: Ib13254f32236a348f575d217253874053acb7b16\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/90f7e706cecdb1f5a8031a22ee72cc950148fd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/90f7e706cecdb1f5a8031a22ee72cc950148fd93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
