)]}'
{"id":"openstack%2Ffreezer~982786","triplet_id":"openstack%2Ffreezer~master~I14a258ead94f4398f98c638f55c227e9e43187a7","project":"openstack/freezer","branch":"master","topic":"bp/centralized-scheduler","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-09 21:20:07.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I14a258ead94f4398f98c638f55c227e9e43187a7","subject":"Register scheduler as central client","status":"MERGED","created":"2026-03-31 13:24:35.000000000","updated":"2026-04-09 22:12:48.000000000","submitted":"2026-04-09 22:11:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982786-bp/centralized-scheduler","meta_rev_id":"51ea553da9ca47449a1661b004bdd5e19870d0d0","_number":982786,"virtual_id_number":982786,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 22:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-09 21:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-09 21:20:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 14:17:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ed18a1065230ff02e072ef2894038acd83ad906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 13:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d085b0879cce705287f06455203fa952e486d2d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:17:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f9b7a6810414d5c8a55ff6dbd4390e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9dd2aadda124a439b2a853ac10abfb6 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf218f0543f3413cb2cba52336961243 : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee1c18d585db4439a6d4314202d22486 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37205745cfc8464f8e07d4fb0b9c62ef : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe650391cc934fd9926cfc33403c822e : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9038361aba304c139d22c99fd069c0ee : SUCCESS in 4m 40s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d74dee827b5641c4a0fc7562c5be47fb : SUCCESS in 33m 35s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/dc4cdda0ff9344648141301daddcc42f : SUCCESS in 22m 13s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f5831f3993f48448ff33d7d2a9084b4 : SUCCESS in 21m 37s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/54ef5048f3b243d9a1dddff468e51865 : FAILURE in 46m 51s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/3766961fb2804d598ec6a013552872f2 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b93b418da8df2e2d7bb6869c9e1d2162f4300c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 15:39:09.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b4847697f59b9d425650f3287379bcdf38a1a7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 15:46:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e57646f7a249b3d6bfce9b57e341c87ff8d02e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:17:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84314a697f88464884238ffcfafa0649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7b3d548ab74ef1ae10281c9f31e71d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bad20738bf4245a61a04784fa8d5d2 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb6131a6de1c4e10a30685e43a33ea1b : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15a0c1a8cbe440b1bb558444d3af6a96 : SUCCESS in 2m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/068b40795ce446699e806c84a029fcb0 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ef27b3ed52c4d9f876644475fbb4ec3 : SUCCESS in 2m 36s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/51930cf102d940ad92f0f7520580a52a : SUCCESS in 18m 00s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/7980e521ac5c413aa689d62d32321a1c : SUCCESS in 15m 01s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f806dccc8be94e4d9cc82a474ce1a588 : SUCCESS in 17m 48s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/a7fc85e2f65f4a15aac151f017f4f056 : FAILURE in 28m 57s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/749ded20e72048bb8bae7bbc322bb6f1 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4f574a53cdca53bf44519ff6f76b1d0f502df92a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-08 18:11:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5b43686d6394b07bb93f2587d27b6967af458da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 18:52:24.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/freezer/+/982160 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"d478ca9345ee8057c256b3fa476ec9b7d4b3140c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 14:59:12.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9c7011b5447a64756cc590953bb5262963625f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 15:46:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05cbb32f78754fb68fb41d8cfe174f9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03d7de3c2234523aa99af86661cd530 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75fb34241aa4e5f9bbbbf0d725a0c9e : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/942f3e2ace344a57afd160a0289d0f8b : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2506939f5b646fc8611c5c931cba0c3 : SUCCESS in 2m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40a87c06a2334a3690dc9b145fb48d96 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d559a417014c43078175408c421ba643 : SUCCESS in 2m 49s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/939e1a437cbb480c87d479c0fce7e620 : SUCCESS in 16m 19s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ac5ba9ae30a743c389e0426ab26264c6 : SUCCESS in 15m 25s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d04e27fd73a64bfaa3edf7fde0526f66 : SUCCESS in 16m 19s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/69e702aa0d0b41d583487d769fd6cbf1 : SUCCESS in 40m 40s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/17397cf153ea4f0aae260349758ebe1e : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":5},{"id":"ba4b80208dfcc4d2da2fb4535cae4e63b11a865f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 21:20:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b1c6104a87cc32ae5a3e3f296a3c824770df2dec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:20:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"23078fe3caaa74131a64574c16f04de53fda27d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 22:11:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/300abed25e8a44d5a78804523bafee2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18268c3715634fc98bc9b895f9944701 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9005357ef79248019cfd55e9b3afbef9 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2143ad7286fa4c37b8f0ce90f12667f3 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33649893408e479ba7fe8af9108601c3 : SUCCESS in 3m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8138c42e60504f709d0d9723ba42fdd8 : SUCCESS in 2m 39s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/3c5bd9c3faa54865ae2a39662d8de18b : SUCCESS in 38m 44s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e5ac14276b1c4c469ebc811d6c67a65a : SUCCESS in 34m 56s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8f53f4a2df8499bb828c2e6511125d3 : SUCCESS in 19m 07s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/bdf8788fc3504512a4a87e899a35f4a7 : SUCCESS in 44m 08s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/811f9d49a27340dc802acab0d59f35c8 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":5},{"id":"1c59fdbd48a7c63830ed902da8579b4a7c4d5f1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 22:11:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"51ea553da9ca47449a1661b004bdd5e19870d0d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 22:12:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3e0ac5bf17436fa80471861eacd02a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e9787553b34ba7b8fc562cc38249b9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b666c58c25d1e83f277a52744c8a5341d07e5f63","revisions":{"721259bf5330a2898f82eea0afd6fe304989c5ec":{"kind":"REWORK","_number":1,"created":"2026-03-31 13:24:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/982786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/86/982786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/86/982786/1"}}},"commit":{"parents":[{"commit":"f6f90cfc5f231c6f13f7cb54f8b5f3f9786dc026","subject":"Add support for trusts auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f6f90cfc5f231c6f13f7cb54f8b5f3f9786dc026"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 13:24:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 13:24:16.000000000","tz":120},"subject":"Register scheduler as central client","message":"Register scheduler as central client\n\nWhen centralized_scheduler is set to True, we need to register the\nclient in the API to be visible for all tenants, otherwise it won\u0027t be\npossible to use the client to run jobs.\n\nThis patch adds logic to pass `is_central` property to the API\nduring registration.\n\nImplements: blueprint centralized-scheduler\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982776\nChange-Id: I14a258ead94f4398f98c638f55c227e9e43187a7\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/721259bf5330a2898f82eea0afd6fe304989c5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/721259bf5330a2898f82eea0afd6fe304989c5ec"}]},"branch":"refs/heads/master"},"d753a273fe427fe0593f6bb21dd4971ea3617639":{"kind":"REWORK","_number":2,"created":"2026-03-31 15:39:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/982786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/86/982786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/86/982786/2"}}},"commit":{"parents":[{"commit":"f6f90cfc5f231c6f13f7cb54f8b5f3f9786dc026","subject":"Add support for trusts auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f6f90cfc5f231c6f13f7cb54f8b5f3f9786dc026"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 13:24:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 15:39:09.000000000","tz":0},"subject":"Register scheduler as central client","message":"Register scheduler as central client\n\nWhen centralized_scheduler is set to True, we need to register the\nclient in the API to be visible for all tenants, otherwise it won\u0027t be\npossible to use the client to run jobs.\n\nThis patch adds logic to pass `is_central` property to the API\nduring registration.\n\nImplements: blueprint centralized-scheduler\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982776\nChange-Id: I14a258ead94f4398f98c638f55c227e9e43187a7\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d753a273fe427fe0593f6bb21dd4971ea3617639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d753a273fe427fe0593f6bb21dd4971ea3617639"}]},"branch":"refs/heads/master"},"4f58bd70e05a1f07dd731eabf7849fd470405bb9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-31 15:46:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/982786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/86/982786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/86/982786/3"}}},"commit":{"parents":[{"commit":"1a90869b476d904b45d124334652d8fb25b0a0c2","subject":"Add support for trusts auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1a90869b476d904b45d124334652d8fb25b0a0c2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 13:24:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 15:46:59.000000000","tz":0},"subject":"Register scheduler as central client","message":"Register scheduler as central client\n\nWhen centralized_scheduler is set to True, we need to register the\nclient in the API to be visible for all tenants, otherwise it won\u0027t be\npossible to use the client to run jobs.\n\nThis patch adds logic to pass `is_central` property to the API\nduring registration.\n\nImplements: blueprint centralized-scheduler\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982776\nChange-Id: I14a258ead94f4398f98c638f55c227e9e43187a7\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4f58bd70e05a1f07dd731eabf7849fd470405bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4f58bd70e05a1f07dd731eabf7849fd470405bb9"}]},"branch":"refs/heads/master","description":"Rebase"},"66e12ab9bc4b730edf7e87485d3b2b61babdff03":{"kind":"NO_CHANGE","_number":4,"created":"2026-04-08 18:11:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/982786/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/86/982786/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/86/982786/4"}}},"commit":{"parents":[{"commit":"8b32a2f302436a6b4354a9fce433ce87f8e433f5","subject":"Add support for trusts auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8b32a2f302436a6b4354a9fce433ce87f8e433f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 13:24:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-08 18:11:04.000000000","tz":0},"subject":"Register scheduler as central client","message":"Register scheduler as central client\n\nWhen centralized_scheduler is set to True, we need to register the\nclient in the API to be visible for all tenants, otherwise it won\u0027t be\npossible to use the client to run jobs.\n\nThis patch adds logic to pass `is_central` property to the API\nduring registration.\n\nImplements: blueprint centralized-scheduler\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982776\nChange-Id: I14a258ead94f4398f98c638f55c227e9e43187a7\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/66e12ab9bc4b730edf7e87485d3b2b61babdff03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/66e12ab9bc4b730edf7e87485d3b2b61babdff03"}]},"branch":"refs/heads/master","description":"Rebase"},"b666c58c25d1e83f277a52744c8a5341d07e5f63":{"kind":"NO_CHANGE","_number":5,"created":"2026-04-09 14:59:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/982786/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/86/982786/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/86/982786/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/86/982786/5"}}},"commit":{"parents":[{"commit":"fc2c479b3e85e4a2b209c807e1e4851414ea1c95","subject":"Add support for trusts auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fc2c479b3e85e4a2b209c807e1e4851414ea1c95"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 13:24:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 14:59:12.000000000","tz":0},"subject":"Register scheduler as central client","message":"Register scheduler as central client\n\nWhen centralized_scheduler is set to True, we need to register the\nclient in the API to be visible for all tenants, otherwise it won\u0027t be\npossible to use the client to run jobs.\n\nThis patch adds logic to pass `is_central` property to the API\nduring registration.\n\nImplements: blueprint centralized-scheduler\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/982776\nChange-Id: I14a258ead94f4398f98c638f55c227e9e43187a7\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b666c58c25d1e83f277a52744c8a5341d07e5f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b666c58c25d1e83f277a52744c8a5341d07e5f63"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
