)]}'
{"id":"openstack%2Ffreezer~982887","triplet_id":"openstack%2Ffreezer~master~I4e1ffa32263d5404d40889ff114d3ced68f7b6be","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-09 21:19:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I4e1ffa32263d5404d40889ff114d3ced68f7b6be","subject":"Switch freezer job to CentOS 10 Stream","status":"MERGED","created":"2026-03-31 17:46:26.000000000","updated":"2026-04-09 22:02:14.000000000","submitted":"2026-04-09 21:59:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982887","meta_rev_id":"a1107bab92652deb2c88d3d2e40c991cc8650885","_number":982887,"virtual_id_number":982887,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 21:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-09 21:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-09 21:19:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 18:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1bb3ba469c9b04dd095f29a2f521679831c353a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 17:46:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93007870f85cbc55dee0e8c1d56a821ce9247332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 18:26:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b0d0c97d7804c0aa338e1211f2d8c25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1fd2a4b6cd340a08ba7dd31b5c50660 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29da57ef15cc411992d80f01b05db0d9 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98719d693e954afc9fbeb55f1b41fc46 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a6660e9945146909e077b52ce5a2c04 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/652f46c36084499eb09414e552129daf : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e815e51c6d604a7ea172544e4612f311 : SUCCESS in 3m 41s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ab1b367e6df542cd9cbcfc0f44a7a51f : SUCCESS in 35m 10s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/53b203810b844651bd3a19e36d914ab1 : SUCCESS in 34m 06s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ce2ccf4322341d3a9ba80efcccf8837 : SUCCESS in 30m 35s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/7f7320e67fde4a488b5f1eac07800249 : SUCCESS in 21m 49s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/4400744a74f3417a82160a03d207f187 : FAILURE in 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"caa219d4c42bd14e47e2e798af0c3bfdffc180f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 13:33:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"46be90f90ca7e286c1e1d615cb142a77977f92af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 14:22:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4123d40549904b99a693ab6f08e2c430\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36aeb15e11e64a3ba988ed596c33f941 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3ad6c24d554d0f92139419df4d68ac : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ad042575d5547bf8c812569719e7566 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b02d3badc89944a09dd841df74562a09 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c777466369b3434cbf926615caeb85aa : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/517f799e85d5436989ea50a3851833e6 : SUCCESS in 4m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/6542f186e4054732b1a3f334abe6291b : SUCCESS in 23m 49s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/85f3054cdff14e6b946e41a02b523e34 : SUCCESS in 32m 50s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ce07d351dc545a4bb32ef9114caced5 : SUCCESS in 34m 21s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/420c0ea7642e4b4f9a437b14f96cfb21 : SUCCESS in 37m 34s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/cf4fe647e1304dc7a6b02287d0398259 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":2},{"id":"7ac2b72a4cf114564c4b91f212444d094277dc4a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 21:19:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"45d85037f2d3ec118bc68a67529f002a2c0b9fa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:20:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bbb74fa10f1420e01f7e0ee3cc382078d3f018b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:59:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90116082b11431bb984dbf1024730a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ff4500d6314475bfd5c41fd738de26 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18bb4fa12a34e95898841fadc58fae1 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd8d0193a8034835a9c661f9eaedc75b : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e20091796cda4075843887f189bd0824 : SUCCESS in 2m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f04d76a91c174d0c865a0356afcc0f87 : SUCCESS in 3m 55s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d82a7b0baa2d4c159bc847db31c8794c : SUCCESS in 18m 42s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/bb8c938958a440b0afcc5ce5f7d7f6a9 : SUCCESS in 28m 30s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8195a9030e6147908d1330f502f5598d : SUCCESS in 36m 01s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/062930c30cc2462e83437168fedee141 : SUCCESS in 22m 40s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/edbe47d98ebe4db1b2c555c3ae865af5 : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2feac0c19f8798b71116db94b2ceb0a05534f316","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:59:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a1107bab92652deb2c88d3d2e40c991cc8650885","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 22:02:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d481f5194564986a51aa1407508a2b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8eba9004a674146bdd9df2f643f3d8d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c458443c7e062e21d5ed5f0a238489874b7cdc25","revisions":{"db91a5b488d2ace0dd3393d5f50b9c2094d42473":{"kind":"REWORK","_number":1,"created":"2026-03-31 17:46:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/982887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/87/982887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/982887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/982887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/982887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/87/982887/1"}}},"commit":{"parents":[{"commit":"90f7e706cecdb1f5a8031a22ee72cc950148fd93","subject":"[doc] Auto-generate CLI options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/90f7e706cecdb1f5a8031a22ee72cc950148fd93"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 17:46:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 17:46:19.000000000","tz":120},"subject":"Switch freezer job to CentOS 10 Stream","message":"Switch freezer job to CentOS 10 Stream\n\nEventually, CentOS 9 job was not doing anything,\nas defined nodeset was not defined correctly for devstack to run.\n\nChange-Id: I4e1ffa32263d5404d40889ff114d3ced68f7b6be\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/db91a5b488d2ace0dd3393d5f50b9c2094d42473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/db91a5b488d2ace0dd3393d5f50b9c2094d42473"}]},"branch":"refs/heads/master"},"c458443c7e062e21d5ed5f0a238489874b7cdc25":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-09 13:33:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/982887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/87/982887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/982887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/982887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/982887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/87/982887/2"}}},"commit":{"parents":[{"commit":"fc2c479b3e85e4a2b209c807e1e4851414ea1c95","subject":"Add support for trusts auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fc2c479b3e85e4a2b209c807e1e4851414ea1c95"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 17:46:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 13:33:01.000000000","tz":0},"subject":"Switch freezer job to CentOS 10 Stream","message":"Switch freezer job to CentOS 10 Stream\n\nEventually, CentOS 9 job was not doing anything,\nas defined nodeset was not defined correctly for devstack to run.\n\nChange-Id: I4e1ffa32263d5404d40889ff114d3ced68f7b6be\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c458443c7e062e21d5ed5f0a238489874b7cdc25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c458443c7e062e21d5ed5f0a238489874b7cdc25"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
