)]}'
{"id":"openstack%2Ffreezer~986012","triplet_id":"openstack%2Ffreezer~master~Ic8d5909690a0ce7794266f3a11dd28debc121a52","project":"openstack/freezer","branch":"master","topic":"bp/centralized-scheduler","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-15 12:11:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ic8d5909690a0ce7794266f3a11dd28debc121a52","subject":"Use fetched user_credentials for agent","status":"MERGED","created":"2026-04-23 20:19:16.000000000","updated":"2026-05-15 13:00:44.000000000","submitted":"2026-05-15 12:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986012-bp/centralized-scheduler","meta_rev_id":"986f85108fb6d9b373eeab6efa9d11368359f903","_number":986012,"virtual_id_number":986012,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-15 12:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-05-15 12:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-15 12:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-23 21:14:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98359292bde2ac145ff50e97cd934213c0095bf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-23 20:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a844f082d39e37bce61600a3325a52a7b64c9619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 21:14:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a1d6a85518549c9b2877b6e19c4763f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515d7163bf1741969837764d591dc368 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93544cb989ad4cb7b7ec14518e8fb61b : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd2f8a773bcc400ba2ccc0f37d73d8f8 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0d038d8cf2f49a1b69cb2abb6bc069b : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b18f79a2e9494eb0b8c9b62da82aeb8e : FAILURE in 6m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e4c2dc38670441695fdca3593a75073 : SUCCESS in 2m 40s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/49edf786df57417db3f1572e6556b5dc : SUCCESS in 46m 11s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e3ba64793def430b993c6ab62252c7b6 : SUCCESS in 32m 22s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9b7f301968b40f3ba7aa938a442d70d : SUCCESS in 35m 21s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/7db503b03f454a0ca330f5416002dce7 : SUCCESS in 37m 38s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/eeddb587debe44ebb4417e711c45254a : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":1},{"id":"da530b3d73b768524d5e3be7fb76a5ed3820129f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 18:18:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"965e168bbd220ee07c4091d48d60574e01d81d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-12 19:01:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6634eb94b40386dbf9a3830b5511a2c139a1f19f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 19:55:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c15457b84a9f48b58e8779836dbc9d83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd466940614f4116b68287b777820903 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30a38f5cec5d49a9bfcb53e8b17ad9ad : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d2e079873440f4bdd6f0a19d5ef336 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d8766352a684c4db4dffc29827f3a83 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4e841e63ee24c4eaf32c07b6013b150 : SUCCESS in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81e084fee79b4aa3b7c62f4f38c859e1 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a44c76fd15db4abc90611307bbc931fa : SUCCESS in 4m 29s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/7ff315c5bac7452184483da49ca83983 : SUCCESS in 21m 34s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/af020429d45c4e3fb4631862fd558a9e : SUCCESS in 14m 49s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fab8393148c45b2905dae79f50e512a : SUCCESS in 25m 49s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/4bcb04c39ba04d81b2fba32ecd27a69e : FAILURE in 27m 44s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/140cd5c7834249f7ae0368c731247f47 : FAILURE in 43m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5df503b7976adc04ca832ed52485dd2cb9a29e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 08:42:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"adec258a13e73e65b0d45b91936d0020ad5c88f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-13 09:10:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5a8fcb356950abac07b9b8a817e1b086fa61c634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:55:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93261373eb0e449bb70a0580c03fe329\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153e4021dcc44238b5cbd405cfaeb9d9 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2a3a84e6394c8d8e14531fbc4c0d80 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c763ac3938a4a429ad77dedccb20702 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/092c35850bde46a39ff2a59e0962ec4d : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e03209aba013480ab8a5709265c1752a : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/17a9f1b09cfe47238df0a44fd7d6289b : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b21ce3f41f584e4d9791e9e989181688 : SUCCESS in 3m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9120e87ee5984048898792bd46b8ff79 : SUCCESS in 35m 55s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/15e6a50144b742cda0ee8d7162259d64 : SUCCESS in 35m 40s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b111f3f3473c4d97a3cfb9ab45bef268 : SUCCESS in 31m 28s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/ae4199caf1744b6c863f1a120b233bce : SUCCESS in 37m 54s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/74b8b339845b4ab0952db54a3e5d5052 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b8a9bd5e67de6fe870611ea9f11187c87d699889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-14 14:55:34.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"dd131d357145144cf898e1646e4110a6e63545d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:49:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91f6af6ea1542558ca212de1dc76c70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305558c9b3c4455aad6d32775632f110 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ad0a6bf780c4ee6a00de79098c2b4f8 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97559ecf86d4966ab77b2d7d7323204 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a513ebdd02754135a3488282bbee471f : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e73f13a0097d47cfa9e06e978c290c75 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59a6dc53165c4205ba6cbe02e54950a6 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8841a23ef8d048509bd663359c23bd12 : SUCCESS in 5m 29s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/25977d7e749d49faa08a59252cf7a39e : SUCCESS in 35m 59s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a5e79bca43fb42af9c3ddee56bb61fa4 : SUCCESS in 37m 03s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e4583bb7d724234a86306f89a52c4d0 : SUCCESS in 24m 08s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/056e04f990024d9a88ecc73f1fd810cb : SUCCESS in 36m 14s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/56deabfeaab643b785a231915c3e1a73 : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":6},{"id":"19863200e1e0295359c2d41d2d20b64e240fb04c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-15 12:11:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4ff4ad19caac570113ad82b5785b9ee37a4c4155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:11:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c219b4c72c02569b42a91bcfb98abf15bac0c5ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:59:41.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d2bf78f59ec4eaab22a4f895b12f8f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4eff0547324b448ad967c27587bb80 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/240e8a1c684144d1bb25440853e5f724 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd6fb1d8fa9499a8f1d4cd3001b89f5 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45342e9259434b7e8c9d75f361cd5b72 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61ab5e824a034235980fe41972665983 : SUCCESS in 2m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4116515721ed439dadd46a80fcaa05ed : SUCCESS in 3m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/067cfaba523e496f92cbcbd22e114443 : SUCCESS in 31m 18s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/2d62e45e74ae45e6b5c49d7514765af1 : SUCCESS in 39m 24s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f3d3e1c163143a4b548d2559deb8e25 : SUCCESS in 36m 13s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/ad90836aeda34d1cb625f23193dca7cc : SUCCESS in 31m 45s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/a8ab6f91013a4dcfb133893b6780fc4d : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":6},{"id":"13e3e0340e0ce616551ba6359a84e1a94a5eaa7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:59:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"986f85108fb6d9b373eeab6efa9d11368359f903","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:00:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe547142fdbd454e9953ef38cab27ed4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eebb881f6784252ae5906a70dc5c51a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bfd246db84481a90df0f9b983f6cea413fdc156d","revisions":{"c1f6f589e6602a98789b66a8cac0bafef71d242d":{"kind":"REWORK","_number":1,"created":"2026-04-23 20:19:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/986012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/12/986012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/12/986012/1"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"subject":"[WIP] Use user_credentials for agent","message":"[WIP] Use user_credentials for agent\n\nChange-Id: Ic8d5909690a0ce7794266f3a11dd28debc121a52\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c1f6f589e6602a98789b66a8cac0bafef71d242d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c1f6f589e6602a98789b66a8cac0bafef71d242d"}]},"branch":"refs/heads/master"},"4958d1dc9ee8af0a197444762f2137ca8b82ab1a":{"kind":"REWORK","_number":2,"created":"2026-05-12 18:18:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/986012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/12/986012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/12/986012/2"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-12 18:17:56.000000000","tz":120},"subject":"[WIP] Use user_credentials for agent","message":"[WIP] Use user_credentials for agent\n\nChange-Id: Ic8d5909690a0ce7794266f3a11dd28debc121a52\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4958d1dc9ee8af0a197444762f2137ca8b82ab1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4958d1dc9ee8af0a197444762f2137ca8b82ab1a"}]},"branch":"refs/heads/master"},"60496b295b257de8368825d382008edac9d4fef4":{"kind":"REWORK","_number":3,"created":"2026-05-12 19:01:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/986012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/12/986012/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/12/986012/3"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-12 19:01:25.000000000","tz":120},"subject":"[WIP] Use user_credentials for agent","message":"[WIP] Use user_credentials for agent\n\nChange-Id: Ic8d5909690a0ce7794266f3a11dd28debc121a52\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/60496b295b257de8368825d382008edac9d4fef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/60496b295b257de8368825d382008edac9d4fef4"}]},"branch":"refs/heads/master"},"3b5ca81269d2555472ccb04900076ef7a125df93":{"kind":"REWORK","_number":4,"created":"2026-05-13 08:42:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/986012/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/12/986012/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/12/986012/4"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-13 08:42:39.000000000","tz":120},"subject":"[WIP] Use user_credentials for agent","message":"[WIP] Use user_credentials for agent\n\nChange-Id: Ic8d5909690a0ce7794266f3a11dd28debc121a52\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3b5ca81269d2555472ccb04900076ef7a125df93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3b5ca81269d2555472ccb04900076ef7a125df93"}]},"branch":"refs/heads/master"},"934e717e9dcdf0dcafcff954de202941b88549a2":{"kind":"REWORK","_number":5,"created":"2026-05-13 09:10:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/986012/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/12/986012/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/12/986012/5"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-13 09:10:07.000000000","tz":120},"subject":"[WIP] Use user_credentials for agent","message":"[WIP] Use user_credentials for agent\n\nChange-Id: Ic8d5909690a0ce7794266f3a11dd28debc121a52\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/934e717e9dcdf0dcafcff954de202941b88549a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/934e717e9dcdf0dcafcff954de202941b88549a2"}]},"branch":"refs/heads/master"},"bfd246db84481a90df0f9b983f6cea413fdc156d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-05-14 14:55:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/986012/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/12/986012/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/12/986012/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/12/986012/6"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-23 20:19:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-14 14:55:34.000000000","tz":0},"subject":"Use fetched user_credentials for agent","message":"Use fetched user_credentials for agent\n\nWhen scheduler is running in central mode, it uses user_credentials\nfor the job execution, which contain credentials for leveraging a trust.\n\nWith that, we need to pass these credentials to the Agent, to actually\nexecute a backup flow with proper project scope.\n\nWe also ensure, that all job metadata update is happening with\nproper scope, by using same user_credentials while updating job state.\n\nChange-Id: Ic8d5909690a0ce7794266f3a11dd28debc121a52\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bfd246db84481a90df0f9b983f6cea413fdc156d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bfd246db84481a90df0f9b983f6cea413fdc156d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
