)]}'
{"id":"openstack%2Ffreezer~986969","triplet_id":"openstack%2Ffreezer~master~I72630514bbb2ed98c12352295faf43197880f40e","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-01 12:07:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-01 12:52:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72630514bbb2ed98c12352295faf43197880f40e","subject":"Remove unnecessary direct dependency on Pygments","status":"MERGED","created":"2026-05-02 11:06:34.000000000","updated":"2026-09-01 12:54:02.000000000","submitted":"2026-09-01 12:52:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986969","meta_rev_id":"0c1f4de1f20112f94b0fea37bb8293ba946dc9fe","_number":986969,"virtual_id_number":986969,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 12:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-01 12:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-01 12:07:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-02 11:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 13:00:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"2984781c48388808d6a6e40d3a4f043c7671f6d9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96b684c93191b3aaa97129da7ea59e367577cb5e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:07:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f1169056f13e5c70f930fabd342d9de7daf45e05","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:08:25.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"bb58ec10a2f7b9f04e586735090c2de536cbfd88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 11:51:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc2cd6318a24b0fb7b22df824e12fb0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7fb76d5f08144039ce585d9f7731ba0 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/59865bfc0d0a48e6a81ced0d7361ff20 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea898873b3e427a9eab5a7c150d3d24 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7348d0b428c41d29ca8acdd86e26549 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3c3aaab3aa5453599c63b7094c30136 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1edf54faf0464a338b4637a104923d0a : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78ad86ff5f0f4fdeaa1eee3424d52da8 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be962883caa84114bc1815487728afd6 : SUCCESS in 3m 57s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f773785fb692401cafff49bb4de06324 : SUCCESS in 35m 24s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/d861b98abfe540449f89ced81237fbba : SUCCESS in 34m 54s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a308aa6a7d84117bcd087497ed3db74 : SUCCESS in 33m 16s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/bf80b0a5d6d24b9793e61d0cff4d2859 : SUCCESS in 38m 35s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/c645c88939494658817319c9955b1c65 : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7aaaa89c1d4ca1e67543bd5a15b0326d05f410d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:39:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5fe60e771593631aefdc49769bf4d401a01d032e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:39:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"22316ed872e4dc3bf11761a414d64a09084d3988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:35:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a54bcce84d44812814b467ec23da2e6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/719460fbe2924ef19a8fe6d57830bbbd : SUCCESS in 3m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16b2435ebe84449a9f15f37ed4747d4b : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a4f6c5452c4853aac382f7dc4e9743 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8530d2cee04a818808c3e556a02378 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7aa6bff48b14820a9d72387389615a1 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73232559702143cbadd18552ff81e6df : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ad8d41877be4a89b66240bf1cdd92e2 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3873cedbea68499b8aa78c3fc42bd79c : SUCCESS in 4m 46s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/17f58090e56b4a3fb64b8a6973a669d8 : SUCCESS in 17m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/c5ab9652e63b49c58e5d77c13d156f93 : SUCCESS in 27m 19s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2531672376634767833cc6fdc782b061 : SUCCESS in 38m 56s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/79197286eb964fdd85f1f68f29e57d9d : SUCCESS in 37m 44s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/5db8efe02bde4af194faf317a8ac82c6 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":4},{"id":"6d9d01ef5c7c6449b2e7386fba1a391a4583d37f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-28 13:00:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8cfaf3d79fe712143ca48bb054b517909e67a6bf","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:23:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"bacf9ccf2902810202a13456532f53d8bc87ab1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:24:02.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7731d478f03e9a8341549a16c5b54fbdff344192","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:24:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"776ef6ff523d4ddf042377bbcb8aced6910de398","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:24:47.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"622d4e7b714795ce02889e657776d7eb4430ba65","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:24:50.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"cae184abe34ebaa0163f970bff5bd05015588f8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:24:55.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"a1e6efff08f842d7554455c19e6538e7804ae30e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:25:19.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c4900bbb207f340719b5c7db0481f4da04ba2f7e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 13:25:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"68f4a1ded8be32aac27cfe636936d0c62343aaaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 14:10:15.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f91f74155cbc4866957fffbfcf1885f5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/efa710c955df4f1986bce4b5c59792cb : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd2a39b068740f8a38db736a1e30e2e : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd73f6247121427281afae500c3a208c : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffd8910bd5c488187bd0fd36964dd5f : FAILURE in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb431cdc76584bf8b066a8962fdaf60b : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af5e9c0d8f1c49dab08f1a1da2b80ab6 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de8df210a7ed4516b51531725a081ef1 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0af9b8f6ea2b49c99106ac5c21c5a00d : SUCCESS in 3m 26s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/020abdbee6874e0d8ada4242cabc6ede : SUCCESS in 25m 38s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/aca5ced7b65c46edabdb99796663da3b : SUCCESS in 37m 18s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbdf828d08d64ae6b16916a5580646c2 : SUCCESS in 23m 06s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/e37389c4925245728f439678d3cb509a : SUCCESS in 40m 37s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/4df25c53151a45878857fc6af01c28b9 : SUCCESS in 34m 41s\n\nWarning:\n  Comments left for invalid file freezer/tests/unit/snapshot/test_lvm.py\n  Comments left for invalid file freezer/tests/unit/utils/test_winutils.py","accounts_in_message":[],"_revision_number":10},{"id":"69b90fb97bbd13c175c9459064f38654ccaae52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:17:24.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"60e34097e01c32db864705b6dbb53cf81e99151b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:58:57.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"96206c72080f47df3a65652be6984ddf3b5ed0f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 04:05:45.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c49a457079d74c7a9d3038f92dc819f8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3350d899f8f24f54a6799e060cf29dd7 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2b34a1386c4fd6af41a702b8ab469b : FAILURE in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a35695952b94f9fa409aa4ede982658 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9cc37cf61dd42538bfc8f142650bf8e : FAILURE in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3c406eba21e4abbaf0b5b618bd5990a : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2b1a25576d54c768f96f3e11d922c1b : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19c9ffa9c46344bea03a26d6e67c6617 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f18f41f2cfd4e07a2a00d2af421389f : SUCCESS in 3m 35s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/587e85192ed8469686384912afd6a27f : SUCCESS in 21m 31s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1e1239258c154db3ad235597767097cd : SUCCESS in 30m 58s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e94a650f5ba48428c2c2a6ade4f8ac9 : SUCCESS in 37m 20s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c38fcfaa7c804f2a8d3d43992d014ae2 : SUCCESS in 32m 49s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/77686a9f9239431ba7259b990ca4e464 : SUCCESS in 36m 47s\n\nWarning:\n  Comments left for invalid file freezer/utils/winutils.py","accounts_in_message":[],"_revision_number":12},{"id":"aa611c1b1ee6320e3d1e7a08c88bf2af640a163e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 12:01:05.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"f23e97c2c73aa272f6af97958387e7a5d2fd8069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 12:01:09.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"6bb595a52b9bc18165973a23d84275b67ded9aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 12:01:30.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"5bb10414510fbae43b5246c8a21375a8c164e4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:54:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6c62e4e4da481b8e09392643cae6ce\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/af4365620658441883cdab580924d401 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c07ca0b1be40648a3d52cd9cd73b14 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777c73ef61f8421fae06267bb2889588 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d44e08ffe04447aca130c4a21284dc : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5afdc320b753445183451fa2527dfa89 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9372ef4bc6a546708dada96a5dd6bb52 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca7a1e570a3b4c12ad77d540fd2cd306 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/165096b0e4c949abb2a774bc74dbc8b6 : SUCCESS in 2m 52s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/1f1537ce69dc4ae6ae5373e97661c41c : SUCCESS in 31m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/d94aae59218048ef8419dd74e848ab69 : SUCCESS in 19m 13s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/268d47b83e8c4358af4730a51abf2631 : SUCCESS in 17m 05s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/39ab6701b795427bb84f56a89547d0d5 : SUCCESS in 39m 55s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/33281e41f037479ca95c9dbf320a949c : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":15},{"id":"e54a7b0b690e4756cbdd377852afd8f0f99df394","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 12:07:29.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"f958ab69b74ac5f99b924b095781db160c9de3bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:07:52.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"331bc454d8874a8689b5b284e0506cc96fe13112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:52:23.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e428bd06aff4e8a896281bd1d0e1082\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/60d7c6e821e6404cbb3f6a3b28f0ac65 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458b28d9b47a4ce4a0e94b9457b99123 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a180b2a3c7492a853d0b67a2248cc2 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748660cda91e4dac930f02b77c516677 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61fd8320a76447ad9ab8a66ca050bbd9 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81214049cc6844f5b880ebc29c656892 : SUCCESS in 4m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/99b4ab1be896411db459a31b2cc60862 : SUCCESS in 4m 04s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/aedc977240bb48a488c37114b35e4361 : SUCCESS in 17m 38s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1ddbd9336a93451dae3a20437b0bc52f : SUCCESS in 39m 20s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05434603ec5b4a6d86c2cfbb640d467a : SUCCESS in 17m 07s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c7caf76dcd7d473581163ce16a9cb2a4 : SUCCESS in 22m 57s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/b8f4255e18e1455784ef3b53d1185dca : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":15},{"id":"493ec5854c531a7da45979888c8fc0dcb749d41a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:52:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"0c1f4de1f20112f94b0fea37bb8293ba946dc9fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:54:02.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4523adf431ed4d8eaab3cc77f3829fc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2416e0efe2941e0909f45820cb743ea : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c9481b39da4e8baf180fba05251515 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"f60d74ee7240940f0d5b5053d3d34cd7af2e991c","revisions":{"0e7557109854de44f1f974ad0f0ca3cfd6363e6e":{"kind":"REWORK","_number":1,"created":"2026-05-02 11:06:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/1"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus).\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0e7557109854de44f1f974ad0f0ca3cfd6363e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0e7557109854de44f1f974ad0f0ca3cfd6363e6e"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"72ac314c84f067fd9c7cfd42f445396de0564fcb":{"kind":"REWORK","_number":2,"created":"2026-05-02 11:07:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/2"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:07:57.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus).\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/72ac314c84f067fd9c7cfd42f445396de0564fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/72ac314c84f067fd9c7cfd42f445396de0564fcb"}]},"branch":"refs/heads/master"},"7af31e0480ada88f0dcd4168143c5a9e65d5229c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-04 16:39:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/3"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 16:39:05.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus) and\nadd Python 3.14.\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7af31e0480ada88f0dcd4168143c5a9e65d5229c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7af31e0480ada88f0dcd4168143c5a9e65d5229c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0dbc45970eb0cd8893c71d7ba95378085ec15ee6":{"kind":"REWORK","_number":4,"created":"2026-05-04 16:39:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/4"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 16:39:26.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus) and\nadd Python 3.14.\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0dbc45970eb0cd8893c71d7ba95378085ec15ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0dbc45970eb0cd8893c71d7ba95378085ec15ee6"}]},"branch":"refs/heads/master"},"7f9849fc5ed600d3d59a825859d068dcbb7b1272":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-08-29 13:24:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/5"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 13:24:02.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7f9849fc5ed600d3d59a825859d068dcbb7b1272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7f9849fc5ed600d3d59a825859d068dcbb7b1272"}]},"branch":"refs/heads/master","description":"Edit commit message"},"34612978a6a012219672ac119ad4359e5749ac12":{"kind":"REWORK","_number":6,"created":"2026-08-29 13:24:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/6"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 13:24:23.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/34612978a6a012219672ac119ad4359e5749ac12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/34612978a6a012219672ac119ad4359e5749ac12"}]},"branch":"refs/heads/master"},"643974eb457cf938544e69b4448e1490b9bbb330":{"kind":"REWORK","_number":7,"created":"2026-08-29 13:24:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/7"}}},"commit":{"parents":[{"commit":"04856451204811bd99ac35713ceff333ae3b93e0","subject":"Merge \"Register scheduler as central client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04856451204811bd99ac35713ceff333ae3b93e0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 13:24:45.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/643974eb457cf938544e69b4448e1490b9bbb330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/643974eb457cf938544e69b4448e1490b9bbb330"}]},"branch":"refs/heads/master"},"adf644d3d442cff96500e381ad6246bc3e9f145e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-08-29 13:24:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/8"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 13:24:50.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/adf644d3d442cff96500e381ad6246bc3e9f145e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/adf644d3d442cff96500e381ad6246bc3e9f145e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"04856451204811bd99ac35713ceff333ae3b93e0","ours":"643974eb457cf938544e69b4448e1490b9bbb330","theirs":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","merge_strategy":"recursive","contains_conflicts":false}},"ad2897c07237c3c226d93923a6530f2ff298c283":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-08-29 13:24:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/9"}}},"commit":{"parents":[{"commit":"dd08c4c64bf23766a8f2296c04ba175736bcd902","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/dd08c4c64bf23766a8f2296c04ba175736bcd902"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 13:24:55.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ad2897c07237c3c226d93923a6530f2ff298c283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ad2897c07237c3c226d93923a6530f2ff298c283"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","ours":"adf644d3d442cff96500e381ad6246bc3e9f145e","theirs":"dd08c4c64bf23766a8f2296c04ba175736bcd902","merge_strategy":"recursive","contains_conflicts":false}},"93e72e4d1e17f1ddc2cb222f7178a3d578456e24":{"kind":"REWORK","_number":10,"created":"2026-08-29 13:25:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/10"}}},"commit":{"parents":[{"commit":"dd08c4c64bf23766a8f2296c04ba175736bcd902","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/dd08c4c64bf23766a8f2296c04ba175736bcd902"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 13:25:18.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/93e72e4d1e17f1ddc2cb222f7178a3d578456e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/93e72e4d1e17f1ddc2cb222f7178a3d578456e24"}]},"branch":"refs/heads/master"},"cd643e9fc916e53624667b688bb7ee5af73db086":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-08-30 02:17:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/11"}}},"commit":{"parents":[{"commit":"5546f601bd0cd654fe9ecb4d8c4220677e2406b0","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5546f601bd0cd654fe9ecb4d8c4220677e2406b0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:17:24.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd643e9fc916e53624667b688bb7ee5af73db086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd643e9fc916e53624667b688bb7ee5af73db086"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"dd08c4c64bf23766a8f2296c04ba175736bcd902","ours":"93e72e4d1e17f1ddc2cb222f7178a3d578456e24","theirs":"5546f601bd0cd654fe9ecb4d8c4220677e2406b0","merge_strategy":"recursive","contains_conflicts":false}},"c58714d6f3323794a5e662168b24a367eb7f01ea":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-08-30 02:58:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/12"}}},"commit":{"parents":[{"commit":"3f270cc556775af544ccdbae23481ba08e0e15bb","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3f270cc556775af544ccdbae23481ba08e0e15bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:58:57.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c58714d6f3323794a5e662168b24a367eb7f01ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c58714d6f3323794a5e662168b24a367eb7f01ea"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5546f601bd0cd654fe9ecb4d8c4220677e2406b0","ours":"cd643e9fc916e53624667b688bb7ee5af73db086","theirs":"3f270cc556775af544ccdbae23481ba08e0e15bb","merge_strategy":"recursive","contains_conflicts":false}},"fd58b203af0f3532765b1c9790b47236e187adc3":{"kind":"REWORK","_number":13,"created":"2026-08-30 12:01:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/13"}}},"commit":{"parents":[{"commit":"3f270cc556775af544ccdbae23481ba08e0e15bb","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3f270cc556775af544ccdbae23481ba08e0e15bb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 12:01:03.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fd58b203af0f3532765b1c9790b47236e187adc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fd58b203af0f3532765b1c9790b47236e187adc3"}]},"branch":"refs/heads/master"},"136d8850ab220777319ca2ce3342d3d8e661ce77":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-08-30 12:01:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/14"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 12:01:09.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/136d8850ab220777319ca2ce3342d3d8e661ce77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/136d8850ab220777319ca2ce3342d3d8e661ce77"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3f270cc556775af544ccdbae23481ba08e0e15bb","ours":"fd58b203af0f3532765b1c9790b47236e187adc3","theirs":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","merge_strategy":"recursive","contains_conflicts":false}},"f60d74ee7240940f0d5b5053d3d34cd7af2e991c":{"kind":"REWORK","_number":15,"created":"2026-08-30 12:01:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/986969/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/986969/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/986969/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/986969/15"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:06:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 12:01:30.000000000","tz":0},"subject":"Remove unnecessary direct dependency on Pygments","message":"Remove unnecessary direct dependency on Pygments\n\nThis library is not directly used, but is installed as a dependency of\nsphinx or doc8.\n\nChange-Id: I72630514bbb2ed98c12352295faf43197880f40e\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f60d74ee7240940f0d5b5053d3d34cd7af2e991c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f60d74ee7240940f0d5b5053d3d34cd7af2e991c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
