)]}'
{"id":"openstack%2Ffreezer~988754","triplet_id":"openstack%2Ffreezer~master~I003834f76b9613a3f3c9ebb88b68e9d86e36b941","project":"openstack/freezer","branch":"master","topic":"bp/centralized-scheduler","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-19 07:18:24.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I003834f76b9613a3f3c9ebb88b68e9d86e36b941","subject":"Fix agent server volume backup logic","status":"MERGED","created":"2026-05-15 11:36:13.000000000","updated":"2026-05-19 07:49:34.000000000","submitted":"2026-05-19 07:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988754-bp/centralized-scheduler","meta_rev_id":"537aecb5f8abc8438245ad034c5374cd3eaacc77","_number":988754,"virtual_id_number":988754,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 07:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-05-19 07:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-19 07:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-15 12:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec2d37adbbeb6ac3b0f25a25343ca894dfe19fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-15 11:36:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8140987ecd6fd980c5807dd5ba76898ee4422535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:29:47.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c15e35eb18634b54889e5b0e08634802\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9942783c035e4cac9fcca475feacf836 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9255b23b59467c81d074e1dcd88f09 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb26f607961e4ef293098a1e5ef78537 : FAILURE in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e1becfbc5c94e85a33f9730407bb774 : FAILURE in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d591689e9a2472cbf3fc4d88081a8bf : FAILURE in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4055a9f603c1448dadcfd7a1d52180d4 : FAILURE in 8m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e96e06e818e540f4bb1410822f4e05dd : SUCCESS in 4m 03s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a22cf9c079ae4b4dbcd3e8dd0186c66d : SUCCESS in 36m 36s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a24ccee6d5a84d95b040b1ee6d621deb : SUCCESS in 34m 34s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9949ae63b0ee4521b282f0da9b9e7eb8 : SUCCESS in 35m 10s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/f0d87127e6f94a068d3165bff8a97b9d : SUCCESS in 39m 42s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/d677002522e2442ab0346949c5fb5924 : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8cb40d6381f59fbc2f035e148207cced444e45a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-15 13:19:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e3e432ac1aade98bb18d9a7f6c05fabd11cb693e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 14:05:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0966370939a64ccbbe4a02b592c566d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b04de08739b403e8f6b53d2401ea8cf : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62681a603d734855a80583fe5f5c6fd9 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e02e070142048eeb12703cdfa6251cb : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d7ac91545144c12af88954456c9deb2 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8525878d72947cab216c6d5af92bd7a : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce9a3a45409d447f8806cc7f6556b10b : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7428199c5544e86b3bd980e08e972fd : SUCCESS in 5m 11s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/6088048a02b84f75b6f4b6f0382e777e : SUCCESS in 34m 53s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b72245c531bf466185f5ca4c210f514b : SUCCESS in 35m 54s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26573f166f5b4dd6ba95e644b3563f14 : SUCCESS in 38m 14s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c074bef7192e400996ab209d9eb08e89 : SUCCESS in 37m 31s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/3ec4b8a48ac648cba22d4f20e3f1f17e : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0b95e2debb08a6a999bf833983cad41bd0b71c21","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-19 07:18:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"882e0e9beaa6244b7640ebbe9c29b2877273a515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:18:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc90206ae91af6a09d036ca743b3d30b1d72cba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:48:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99dc9e7889664177aacf9887d7599fb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783b420395184909ba4985802c758e23 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19b4fcb5f59c49068fdb4d8f28c5f596 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbfbeb13f915472cba6f84552f290bd9 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e6f127a41644084b61bf2adb385cc3b : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/229eaa1ce26a4c9691609467a55710ac : SUCCESS in 2m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/467da0cfd943430fb5da78dd1672bb94 : SUCCESS in 3m 15s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/c8e39d1769bc4c19a24f619d2dbb795f : SUCCESS in 15m 24s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/bbf49d5406534b5590852a1ab459c521 : SUCCESS in 18m 57s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5012bd60c61346da8baa7214ecfbb6cd : SUCCESS in 15m 17s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/810867c7bba54d56b64d0c5c742c2fb0 : SUCCESS in 23m 44s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/fe01e63c675145f0820be2507ae5b36e : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8e9c81cefb282144085320ed71ba677837957840","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:48:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"537aecb5f8abc8438245ad034c5374cd3eaacc77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:49:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/232a82638bd94489af3b5ac8902b1d75\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417f376a8c234387ae2dba5b8087e4db : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2321910f3d05ef50ea35871e440f065838ef778","revisions":{"33a1388f66d34f803f59c11724d6f1cc07e60a6c":{"kind":"REWORK","_number":1,"created":"2026-05-15 11:36:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/988754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/54/988754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/988754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/988754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/988754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/54/988754/1"}}},"commit":{"parents":[{"commit":"bfd246db84481a90df0f9b983f6cea413fdc156d","subject":"Use fetched user_credentials for agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bfd246db84481a90df0f9b983f6cea413fdc156d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 11:35:57.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 11:35:57.000000000","tz":120},"subject":"Fix agent server volume backup logic","message":"Fix agent server volume backup logic\n\nWhen we are iterating over cinder volumes attached\nto the instance, we don\u0027t want to return after the first occurance, but\nrather proceed with cleanup for all volumes before returning.\n\nWe also make logging more clear for scheduler,\nand base error/warning LOG messages based of the\nagent exit code.\n\nChange-Id: I003834f76b9613a3f3c9ebb88b68e9d86e36b941\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/33a1388f66d34f803f59c11724d6f1cc07e60a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/33a1388f66d34f803f59c11724d6f1cc07e60a6c"}]},"branch":"refs/heads/master"},"d2321910f3d05ef50ea35871e440f065838ef778":{"kind":"REWORK","_number":2,"created":"2026-05-15 13:19:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/988754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/54/988754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/988754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/988754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/988754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/54/988754/2"}}},"commit":{"parents":[{"commit":"bfd246db84481a90df0f9b983f6cea413fdc156d","subject":"Use fetched user_credentials for agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bfd246db84481a90df0f9b983f6cea413fdc156d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 11:35:57.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 13:19:01.000000000","tz":120},"subject":"Fix agent server volume backup logic","message":"Fix agent server volume backup logic\n\nWhen we are iterating over cinder volumes attached\nto the instance, we don\u0027t want to return after the first occurance, but\nrather proceed with cleanup for all volumes before returning.\n\nWe also make logging more clear for scheduler,\nand base error/warning LOG messages based of the\nagent exit code.\n\nChange-Id: I003834f76b9613a3f3c9ebb88b68e9d86e36b941\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2321910f3d05ef50ea35871e440f065838ef778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2321910f3d05ef50ea35871e440f065838ef778"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
