)]}'
{"id":"openstack%2Ffreezer~988757","triplet_id":"openstack%2Ffreezer~master~I5f95e928c820f56c282406528592a84f2dad0574","project":"openstack/freezer","branch":"master","topic":"bp/centralized-scheduler","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-19 07:18:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I5f95e928c820f56c282406528592a84f2dad0574","subject":"Fix invalid escape sequence warning","status":"MERGED","created":"2026-05-15 12:37:52.000000000","updated":"2026-05-19 08:05:45.000000000","submitted":"2026-05-19 08:04:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988757-bp/centralized-scheduler","meta_rev_id":"364b2f6ad3f403b830d73a91f3fffcf7d7d32bc5","_number":988757,"virtual_id_number":988757,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 08:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-05-19 07:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-19 07:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-15 13:16:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"635a7e47891ad1a37bd15e4b7106bcc1388263bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-15 12:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ff6eda7cde724c4030823c49225561a25956e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:16:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3698ae3856f04d999276c6af75ba57ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16cd14319ef4528bff88878c5b3d683 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5923d04d77964c8989fcc92edb2ab519 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba91641111a472e88319dadbf543bdf : FAILURE in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19ba3972dcc24914825c520def2ed93e : FAILURE in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c200280584b4221a56035421ce197df : FAILURE in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc38d742119e4f0c90346d0be4fae493 : FAILURE in 3m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/999e21b9206c4ab4af0b0a415f616f4a : SUCCESS in 2m 15s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/20606360880f4027ba42c5b573056807 : SUCCESS in 15m 37s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/9d5ce2b66b124f1ba49229ed75d57aa7 : SUCCESS in 30m 32s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cf0570cf7d94f9c9c3bc5c912b80af0 : SUCCESS in 36m 58s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/549664e2f2c4450e96244db3889c5c2e : SUCCESS in 18m 38s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/9fa86cc885df4c448deb8a2a82511073 : SUCCESS in 30m 43s\n\nWarning:\n  Comments left for invalid file freezer/scheduler/scheduler_job.py","accounts_in_message":[],"_revision_number":1},{"id":"dea61411cf64bd4dce8a249baa41f1d172bfdc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-15 13:19:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f90a595ba6a834819a76a233f2570aabb32b1577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 14:10:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ecbe1a17f644bcc89f470eb25b4c5a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf3db9457a844eea50465987157eac0 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7797e84df3e4a1684533a0a622b1f1b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a848c587679b419caf9ad97e79d4c7f4 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50eb50b475724dfe846a4718faa8f03d : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0bec21f0527479d898451d64c85793e : SUCCESS in 5m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e33b5e2dec994991901fd4c3dbce7f18 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce1e7510265f473a9b1570a9fc180390 : SUCCESS in 4m 27s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b72d86df2c9245ba99c29e276ed6d308 : SUCCESS in 36m 53s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/23d7814675f44055a41e0f9325d72c8b : SUCCESS in 37m 44s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cd13627c5aa44bbb78ae263943e7666 : SUCCESS in 28m 57s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/d01cee7ed3bf48d4bec41cb1a2390c64 : SUCCESS in 22m 36s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/9347c5ac76c84b45839e0408560a6836 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0aeca9d073af84d1c8145c4cd0c10dbeb564c995","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-19 07:18:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"69dcaaf696353d2991789a6eaeb075ec9c26f20f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 07:19:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0377b178ff897059d5edd9ada1a3aae5c7c94ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:04:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b1dc23b660426d85d5c483ccf3b3b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6ed847317c49579c3852e000102181 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/568a4a59c04443cfac3ee7bb0648b7c9 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22e88ee45254878816406f456de6a7f : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24d1f085b33441bda6b5ef653f7efed7 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cb86c2fd8c44589adbbffcff486112e : SUCCESS in 2m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca9316344a1945fc8c57180e8cffa802 : SUCCESS in 3m 58s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/59ec3fdd9d5c4f3da07fdd62e265656a : SUCCESS in 36m 05s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/9376168eed2d41b299d293f91c3bc806 : SUCCESS in 27m 12s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac36c1ff6da4a64a170855ece90a3bf : SUCCESS in 19m 08s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/873047b23002444c89f2fdf085bb5429 : SUCCESS in 38m 22s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/1bb376a942fe44848d02764f973ac882 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4f1f459032c7135f2ce607e938ce2050fd90ec8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:04:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"364b2f6ad3f403b830d73a91f3fffcf7d7d32bc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:05:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/454356cd8d414968aa952a226e3c4979\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e07b758437e41f0a2a11deed9bf169b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"509f5b8860c2d3e18526bb527da5163dd59f58d3","revisions":{"6b7e70232e057623c36b34176ec74e5a3ba0ec4d":{"kind":"REWORK","_number":1,"created":"2026-05-15 12:37:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/988757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/57/988757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/57/988757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/57/988757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/57/988757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/57/988757/1"}}},"commit":{"parents":[{"commit":"33a1388f66d34f803f59c11724d6f1cc07e60a6c","subject":"Fix agent server volume backup logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/33a1388f66d34f803f59c11724d6f1cc07e60a6c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 12:37:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 12:37:30.000000000","tz":120},"subject":"Fix invalid escape sequence warning","message":"Fix invalid escape sequence warning\n\nCurrently, the doc string for the vss_create_shadow_copy raises the\nlog record `SyntaxWarning: invalid escape sequence \\\u0027\\\\S\\\u0027\\n  \"\"\"\\n\u0027\n\nThis is due to to windows-like path in the doc string.\nLet\u0027s ensure that log string is raw-like to avoid issue from being\nlogged.\n\nChange-Id: I5f95e928c820f56c282406528592a84f2dad0574\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6b7e70232e057623c36b34176ec74e5a3ba0ec4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6b7e70232e057623c36b34176ec74e5a3ba0ec4d"}]},"branch":"refs/heads/master"},"509f5b8860c2d3e18526bb527da5163dd59f58d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-15 13:19:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/988757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/57/988757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/57/988757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/57/988757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/57/988757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/57/988757/2"}}},"commit":{"parents":[{"commit":"d2321910f3d05ef50ea35871e440f065838ef778","subject":"Fix agent server volume backup logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2321910f3d05ef50ea35871e440f065838ef778"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 12:37:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-15 13:19:27.000000000","tz":0},"subject":"Fix invalid escape sequence warning","message":"Fix invalid escape sequence warning\n\nCurrently, the doc string for the vss_create_shadow_copy raises the\nlog record `SyntaxWarning: invalid escape sequence \\\u0027\\\\S\\\u0027\\n  \"\"\"\\n\u0027\n\nThis is due to to windows-like path in the doc string.\nLet\u0027s ensure that log string is raw-like to avoid issue from being\nlogged.\n\nChange-Id: I5f95e928c820f56c282406528592a84f2dad0574\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/509f5b8860c2d3e18526bb527da5163dd59f58d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/509f5b8860c2d3e18526bb527da5163dd59f58d3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"6b7e70232e057623c36b34176ec74e5a3ba0ec4d","theirs":"d2321910f3d05ef50ea35871e440f065838ef778","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
