)]}'
{"id":"openstack%2Ffreezer~993405","triplet_id":"openstack%2Ffreezer~master~I4626ee28cec6fc1a0d8787817d8b3d7c30456f43","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-16 10:57:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I4626ee28cec6fc1a0d8787817d8b3d7c30456f43","subject":"Fix SDK usage for cinderbackup","status":"MERGED","created":"2026-06-15 16:11:17.000000000","updated":"2026-06-16 11:43:12.000000000","submitted":"2026-06-16 11:42:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993405","meta_rev_id":"b8743564c3dbe76d262ad39df5f7190c839fc2a0","_number":993405,"virtual_id_number":993405,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 11:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-16 10:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-16 10:57:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-15 17:03:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9038c602347bdbc02c26caa390c3d6cb135ad705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-15 16:11:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea252665d4d5aa7f32a2e964861b479f51c66b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 17:03:01.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/625fdf9a843241ebad7a9dc507fad782\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961cc81b1498445db0d673d33f12284b : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b7e9af8ac794c6094f90ae5ee96b1ac : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e721acc6f4ce45af8a8b608903259447 : FAILURE in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c80bbbc8b7641b7aa3d09d76019e61c : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8dc001b34af45cb916c1668d6d2b98e : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4777df73ad1049d9978593f7dfca10eb : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be7bdd6bd4f44084b2542d4dd4f01c64 : SUCCESS in 4m 11s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/c5593402cbbe4c6f95222e154abb83cd : SUCCESS in 37m 23s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/4a58ae7b10af4c8ab9ed23cd3fdd9e90 : SUCCESS in 21m 07s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de719e530c69451eb0df9699677ae48a : SUCCESS in 40m 17s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c941911821994021a48d3148dc6c1cc5 : SUCCESS in 36m 08s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/67edf77689904387aa2061217731768e : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":1},{"id":"38be446bc30094e2ea6d1ee42a7fb8b10f0c0435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 07:55:01.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"59874d4a483652a8ba86e47c5e3c06e6ab4b4edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:45:45.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91b307cc9fc54dc383457aeadf438d82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d18beb4c4994a9dba9edbfbcdb7300d : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/304aed0b07944e70b74a320b22c045d9 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8ac4ad6fad49d6bdb97dea5185cb6f : FAILURE in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b9364dc8fce4e08b1b8cb21881aba02 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5f24528475242019ccf02cdd7b35df7 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ba3a8c5dbd3478fa1aeeb0257408ecc : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83d64b54039c452094b4eda09b003a7d : SUCCESS in 3m 44s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/71f4d444069e4f21b69d6f566141c594 : SUCCESS in 20m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3aa831db221f4674ade81fd23b2f03db : SUCCESS in 19m 48s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3c04f635d6248e7aed143ad6c82f946 : SUCCESS in 39m 25s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/87c4ea473b3a48958f76443fb3fd53da : SUCCESS in 45m 18s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/bf0c1748ab594ebfba54fb5435b99621 : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":2},{"id":"24e882c12d8ebc270a28498077a35b596cca923c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 08:58:11.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"603a064674914ac540cda9007caba5cac2e1bfc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 09:59:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120b123d4ec3457ab1a321042c4d4689\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f09657958474cadb42811c45269d85b : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/192c6f4607f747419c1abbf6c1ac8353 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf0872bd7274a999e7335011d3acae8 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca89ef78546f4cb7ad5bf85c68e8265b : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a71be7604424e8d992a57402599418e : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c1aed441a1a4421b672eb873ab26630 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b80866d7d3124fb4aeacbae477df96e6 : SUCCESS in 5m 09s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/3a4b95ee5d894b67b6fb09263be26f77 : SUCCESS in 35m 40s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/558ae00cdea146c19b48ec21816d7c6c : SUCCESS in 35m 06s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74e689924eda4e4da22863c6e0cf8f0d : SUCCESS in 38m 41s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/179de630d94d4d7095043f8f6180b780 : SUCCESS in 40m 40s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/126a8b609fbf4e5ab5b34b74fecc8bda : SUCCESS in 52m 37s","accounts_in_message":[],"_revision_number":3},{"id":"53049a440233d6b5afe7e9f61faaead30b64e2cb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 10:57:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ebf41a440153301f757cccb6fa5db89729a50199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 10:57:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee3c876c3b68034a4489be39ec42ceeeea2f94c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:42:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b381c0ac5684c908636e615659ac906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af24e320fba348e4afdc5fbb01ebdb3a : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5339327aa804088a2c679a975b6dd42 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8574463eea498595622a2989cd5ff4 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f5d6d51dcfa457892fc26c1570bd636 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4ca9c13336f4cd0a60a027b75dc9e9e : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/69460df6184b49df88d420177942f6ad : SUCCESS in 3m 52s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e07e0920a8ef4505bed05c298bf33d1c : SUCCESS in 34m 29s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/00a695e8c26c4313ae449e29fb45c7cc : SUCCESS in 34m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b54751f86da4a1791aa810ef7074088 : SUCCESS in 35m 29s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/f1c2e80fb65043848d9207c2729a2f0b : SUCCESS in 39m 13s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/b04e7497d1124389964184afd8713d27 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1ff2144f829f86c8da9e58d653c9fc9a64f12c61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:42:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b8743564c3dbe76d262ad39df5f7190c839fc2a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:43:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc27701d4e4492da7c4e2cb7392a33b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9122ca1bfd0142f0a864eb98df80a029 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16c11f68fb223968ce2011b2a9f699cf71dac938","revisions":{"a29a996c5601213ad8d079cc89406c3f189dbc95":{"kind":"REWORK","_number":1,"created":"2026-06-15 16:11:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/993405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/05/993405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/05/993405/1"}}},"commit":{"parents":[{"commit":"c6b6c004bf65df35ba604145a61411124bd0215e","subject":"Add support for container naming to cindernative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c6b6c004bf65df35ba604145a61411124bd0215e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-15 14:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-15 14:12:23.000000000","tz":120},"subject":"Fix SDK usage for cinderbackup","message":"Fix SDK usage for cinderbackup\n\nOpenStack SDK returns a generators when we get a list of backups.\nWith that the agent treats it as a list historically, which completely\nbreaks admin jobs, as len() of generator will result as false.\n\nWe also generally refactor admin ops for consistency and to reduce\nrepeating simmilar code.\n\nChange-Id: I4626ee28cec6fc1a0d8787817d8b3d7c30456f43\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a29a996c5601213ad8d079cc89406c3f189dbc95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a29a996c5601213ad8d079cc89406c3f189dbc95"}]},"branch":"refs/heads/master"},"14306cf051b1eddc801cb5c1ef23ed8a249daf34":{"kind":"REWORK","_number":2,"created":"2026-06-16 07:55:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/993405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/05/993405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/05/993405/2"}}},"commit":{"parents":[{"commit":"c6b6c004bf65df35ba604145a61411124bd0215e","subject":"Add support for container naming to cindernative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c6b6c004bf65df35ba604145a61411124bd0215e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-15 14:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 07:55:01.000000000","tz":0},"subject":"Fix SDK usage for cinderbackup","message":"Fix SDK usage for cinderbackup\n\nOpenStack SDK returns a generators when we get a list of backups.\nWith that the agent treats it as a list historically, which completely\nbreaks admin jobs, as len() of generator will result as false.\n\nWe also generally refactor admin ops for consistency and to reduce\nrepeating simmilar code.\n\nChange-Id: I4626ee28cec6fc1a0d8787817d8b3d7c30456f43\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/14306cf051b1eddc801cb5c1ef23ed8a249daf34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/14306cf051b1eddc801cb5c1ef23ed8a249daf34"}]},"branch":"refs/heads/master"},"16c11f68fb223968ce2011b2a9f699cf71dac938":{"kind":"REWORK","_number":3,"created":"2026-06-16 08:58:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/993405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/05/993405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/05/993405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/05/993405/3"}}},"commit":{"parents":[{"commit":"c6b6c004bf65df35ba604145a61411124bd0215e","subject":"Add support for container naming to cindernative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c6b6c004bf65df35ba604145a61411124bd0215e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-15 14:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 08:58:11.000000000","tz":0},"subject":"Fix SDK usage for cinderbackup","message":"Fix SDK usage for cinderbackup\n\nOpenStack SDK returns a generators when we get a list of backups.\nWith that the agent treats it as a list historically, which completely\nbreaks admin jobs, as len() of generator will result as false.\n\nWe also generally refactor admin ops for consistency and to reduce\nrepeating simmilar code.\n\nChange-Id: I4626ee28cec6fc1a0d8787817d8b3d7c30456f43\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/16c11f68fb223968ce2011b2a9f699cf71dac938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/16c11f68fb223968ce2011b2a9f699cf71dac938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
