)]}'
{"id":"openstack%2Ffreezer~995966","triplet_id":"openstack%2Ffreezer~master~I92d14b08bec596c47931d5947d5d8680bfe27dcc","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-07-07 11:40:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I92d14b08bec596c47931d5947d5d8680bfe27dcc","subject":"Fix scheduler underscored arguments","status":"MERGED","created":"2026-07-03 14:51:20.000000000","updated":"2026-07-07 15:48:54.000000000","submitted":"2026-07-07 15:47:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995966","meta_rev_id":"d511fcd42dbd26cd5a9ac6f635f9c276c68e482d","_number":995966,"virtual_id_number":995966,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 15:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-07-07 14:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-07-07 14:27:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-03 16:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a828f399eddd0959a65309fc2635477d219538f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-03 14:51:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c4ecd87470997d5329246e77b0b84da8010d40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 16:34:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/509f8435eb044d009217ba8e24579453\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd21c6bab674344907c124e53fafc4a : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fda5a59048864b3493d2183779a3cc4e : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f444011167448219d0f0f6af89b02c7 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6b38a11cf3442ceb8deb7c153cffcbf : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51214c676d6740b79a8020573a26bffe : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5499484595834cb4b86ee643a9016314 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa43cb49f3af40dcaa56e85fbfff6fe1 : FAILURE in 5m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/270c820187d8410f92ea418eeca2bb4f : SUCCESS in 36m 05s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ebe3b5c378e141d1a8cf5043495df172 : SUCCESS in 25m 50s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/234d6a817611468792fff9eb218b8756 : SUCCESS in 36m 01s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/eeb8e9abd1734edcab04f54be7debac6 : SUCCESS in 41m 35s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/767938e91e004a6b88ecc772259ecc2c : SUCCESS in 1h 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ca9b631d123e72702eb3eeb020f3ede5dd88edfc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-07 11:40:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"391df2f661e1811668a39454fd14d6a8323c5e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 13:50:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f034c0ec9d4d6a994cae441c02072a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd1995496bf4f53bfe96c8d48856f2a : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b18c9764e8f343cf8362dae0c882f85d : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ea82b8a01e4f6f9928fdde7eb52656 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/023b21dd23fa462180affcc45d231d35 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2734fb6e80e44f3ae38875296fd5d38 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05e852bff15140ff8d5a9a35f762e04c : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/676577d512394615957bbc414990c7ff : SUCCESS in 4m 22s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ef592e626fa147c4a95052e464bb05e7 : SUCCESS in 37m 28s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ae9f8539f17a4e878144f4c3cf6917c7 : SUCCESS in 37m 25s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9108e664e8a44416b5ed128e0b5c3da4 : SUCCESS in 33m 52s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/acb260d73bf14cc28fb56f11f46ca36a : SUCCESS in 43m 46s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/3e9f9ff2cb50434496338ba9eb5459a0 : SUCCESS in 2h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b3dbc1f225c1733c7c40a6907dba1825789af678","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-07 14:27:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c44a78d46daaf7ab694bf17dcceee0010521543f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 14:27:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"762b50e1a0507246099e4a7c07710b7cde4124b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:47:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b64d0c2e4dd74cdf89914c5bb4742534\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c80ded99bc44f98bc408cf9d17e6d70 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e12bc5667dad4e9091655160ee8bc861 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44b5693f5444a31823ed29b8c4ed882 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94209040a0684237ac798056e6172781 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bb2ca31f6534c6c885a03b4bf511017 : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e3bc735e03742cd9880934097864a07 : SUCCESS in 4m 12s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9fff1e0dc5a249bc9f916df9716569fd : SUCCESS in 39m 59s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/aa51aee0de054f2e8dedcd6a2838e3ff : SUCCESS in 33m 22s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d636857528746b4954d328a1b23fc5d : SUCCESS in 41m 59s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/0e3a52f6582e437d916d0f0c12c729ed : SUCCESS in 28m 27s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/f39f163939314554978f849bbf490d31 : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"93dcab7e6cb886fc82b0d989a734c13bf7c3c18d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:47:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d511fcd42dbd26cd5a9ac6f635f9c276c68e482d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:48:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008c08104c8c486ba03545a60df6c0cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6326f7053240c1a3adec18b42b2933 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40ab1d935c97c3204f3a97a7f9d9e0a587df9b8a","revisions":{"40ab1d935c97c3204f3a97a7f9d9e0a587df9b8a":{"kind":"REWORK","_number":1,"created":"2026-07-03 14:51:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/66/995966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/66/995966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/995966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/995966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/995966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/66/995966/1"}}},"commit":{"parents":[{"commit":"5abb0319e1cd308270baf1cdfd8d1cdfe0d235ee","subject":"Merge \"Implement dynamic backup name templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5abb0319e1cd308270baf1cdfd8d1cdfe0d235ee"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-03 14:51:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-03 14:51:12.000000000","tz":120},"subject":"Fix scheduler underscored arguments","message":"Fix scheduler underscored arguments\n\nSome arguments were underscored, rather then\ndash-separated, which is confusing and misleading.\n\nWe align arguments naming here.\n\nChange-Id: I92d14b08bec596c47931d5947d5d8680bfe27dcc\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/40ab1d935c97c3204f3a97a7f9d9e0a587df9b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/40ab1d935c97c3204f3a97a7f9d9e0a587df9b8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
