)]}'
{"id":"openstack%2Ffreezer~996018","triplet_id":"openstack%2Ffreezer~master~I290c23513396674a2a17c733e9bb41bd3c958571","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I290c23513396674a2a17c733e9bb41bd3c958571","subject":"Don\u0027t use v1 client","status":"MERGED","created":"2026-07-04 17:06:33.000000000","updated":"2026-07-07 11:18:25.000000000","submitted":"2026-07-07 11:17:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996018","meta_rev_id":"db0d38051912e61e0d82a05525c180027097e6f7","_number":996018,"virtual_id_number":996018,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-07 11:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-07-07 09:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-07-07 09:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-04 19:23:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c1baddc20aa7b434299a3f3c6429272e0de0b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-04 17:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c8b6e66bf31e5139b1f87801481d1f930334b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 19:23:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668397c3ec9c4aed8bac20fa015f2b56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc59ae65fa1446ea48af00366c45b11 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d53cbecb09364dda9699fa3b2232dd83 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5b22be15094a13ad718ef6af7511c3 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c9e567c62a0435585c90a5acc97474a : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf892b925fe4482be39249fae40ca73 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/232e486b15834b0cbae37ae0df3c7d81 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/151bab4c261e43bd9c9ae1ddd288ff91 : SUCCESS in 4m 05s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b97a8f4629304ea7a015a05c2bc1c3b6 : SUCCESS in 38m 11s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/78631457c6b84bf5a7b4c792fab1f2de : SUCCESS in 34m 04s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0d9a5cc36994ef0b895092adbe68960 : SUCCESS in 35m 42s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/1ae43d1855224846a992423bb156086b : SUCCESS in 40m 30s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c2e070cdf3d4dd19a5bff0fa70036b9 : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c1f3eb420eb681b72bc2df2559e47bb28146580f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-07 09:53:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebd6dadb5bc38edddfce0d3ba003c2b28eaed298","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 09:54:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bbf3bdbb6591c8e8c826a298249c057188168f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 11:17:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c0e10e9a26f4c8b9f92cc5e8e3d3201\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807c0a6186a54c21baccc7b77584708a : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a39e4992f6244111a8be35c9a292834a : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4fb58564cb040658ea407b2d9f71f84 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12fe671c287f4460bfb5576fa143fb55 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdd398934950405391f4a930134dd80e : SUCCESS in 4m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/750f4149720741569de1f0437645cb1b : SUCCESS in 5m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/79b6fdf87ee34a94b97d69ac3a63d7f7 : SUCCESS in 40m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/76ea54a89b6d4a85b45160432067101a : SUCCESS in 36m 07s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42e0d44af99746a791fe5809d84f22fd : SUCCESS in 34m 45s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/0b0b7cae44354f72a1f0b738a37ef8bc : SUCCESS in 40m 06s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/315de3f6511f49678dbe64c262bd64ab : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0c9d63a536605995f50097c91b20c9a3d611615d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 11:17:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db0d38051912e61e0d82a05525c180027097e6f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 11:18:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f54f71b53d3d468ea1845981bca3edce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725884bc4ce047528858222ce74577b5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ca607282a37dd5b5e76dc116595ff55fecfa9cb","revisions":{"6ca607282a37dd5b5e76dc116595ff55fecfa9cb":{"kind":"REWORK","_number":1,"created":"2026-07-04 17:06:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/996018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/18/996018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/18/996018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/18/996018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/18/996018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/18/996018/1"}}},"commit":{"parents":[{"commit":"5abb0319e1cd308270baf1cdfd8d1cdfe0d235ee","subject":"Merge \"Implement dynamic backup name templating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5abb0319e1cd308270baf1cdfd8d1cdfe0d235ee"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-04 17:06:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-04 17:06:26.000000000","tz":120},"subject":"Don\u0027t use v1 client","message":"Don\u0027t use v1 client\n\nv1 client has been removed from python-freezerclient in favor of v2.\nWe should not attempt using it.\n\nChange-Id: I290c23513396674a2a17c733e9bb41bd3c958571\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6ca607282a37dd5b5e76dc116595ff55fecfa9cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6ca607282a37dd5b5e76dc116595ff55fecfa9cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
