)]}'
{"id":"openstack%2Ffreezer~998624","triplet_id":"openstack%2Ffreezer~master~I30f2c8ec6444c9e00a33c770eb269f53fd41d114","project":"openstack/freezer","branch":"master","topic":"freezer/backups_update","attention_set":{},"removed_from_attention_set":{"39044":{"account":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"last_update":"2026-08-18 11:06:19.000000000","reason":"\u003cGERRIT_ACCOUNT_39044\u003e replied on the change","reason_account":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-18 11:14:20.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I30f2c8ec6444c9e00a33c770eb269f53fd41d114","subject":"Avoid rotation of manually created cinder-backups","status":"MERGED","created":"2026-07-24 14:06:15.000000000","updated":"2026-08-18 12:00:36.000000000","submitted":"2026-08-18 11:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998624-freezer/backups_update","meta_rev_id":"ece958bac27c7180e809cf7b6f3c14e02f1aef8d","_number":998624,"virtual_id_number":998624,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-18 11:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"all":[{"value":1,"date":"2026-08-18 11:06:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":2,"date":"2026-08-18 11:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":1,"date":"2026-08-18 11:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-24 15:22:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-28 12:39:01.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"CC"},{"updated":"2026-08-18 11:06:49.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"REVIEWER"}],"messages":[{"id":"be64c1d96d05bcee7ec892a3252d84ca6ce0983f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-24 14:06:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3cd85d8486ee8b969ea496030e79f0264b8e159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 15:22:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31350f2acee6416ba6fadf3b1041b109\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe88efd49a3844af9d97bb453edbdcc7 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bd375d3719b471abaae6585a8d5b6e6 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d345dedec64dc3807fdbc46025634b : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/872bd7f453e645dda38e540fdb648244 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7349328f6a9460e9b05a4eaba2a2d7d : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a72d398fa15418e98e6864f680a6dec : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f70173cbe03049da82081408ce23c29b : SUCCESS in 4m 07s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/02ba5d7da2e148fe8f85b66b82c3a48b : SUCCESS in 41m 58s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/bd67c791eaf24ddebef87864151c6d7b : SUCCESS in 39m 46s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/876e280c3193488bbebef472d5af96ce : SUCCESS in 40m 22s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/6baa83a257454b99a584f0a52be7185f : SUCCESS in 39m 58s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/807f2ee1365e47e3aff55271f948b016 : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f54e04572820afa1172fadd3839e89f3bef06971","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-07-28 12:39:01.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"21150c33e24c7a17af866df50d44ac2e274e7e24","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-29 08:47:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db1808ffd88d2b581610119f6e779a0a48b75ea3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-29 08:58:45.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f542d094a949682c869ee370507f300f146345e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-17 12:50:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fcad45380aaa4441a4b2fc53164861db5a098605","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-17 13:25:35.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8675f025ac933ffa998fce449df401769029154f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 13:34:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b86a3dd819c40698036bed060629334\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48af169915d4ec1bbbbffc4de63a882 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b6dd538fe574337a8154a7eab6930b9 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca362e13f6c46ef937cf0e8e16d6ae1 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a872a3514baa4ce0bc58193913210821 : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/babaf0bd715c46c6b4f718d34db55094 : SUCCESS in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c43af753b930407290da22e1da136ca7 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a790c68dff44f83acc67710181a610f : SUCCESS in 5m 05s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9cd50c2718e14839ad1be4a581f6e87d : SUCCESS in 32m 03s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/eb26405a174f452e9996d7f723ea5242 : SUCCESS in 38m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/846f4bc3eb0c4e0bbfa8d7db38ed349e : SUCCESS in 16m 23s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/6a3c1ab2ad48487faf39f72728ee86cc : SUCCESS in 31m 53s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/33565e98a7b54163a0b9b7efadad5321 : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":2},{"id":"27ec86224b7b2d645fe11c83b11da967b3659484","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-08-18 11:06:19.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"60bfaf25f2f7b9fcab80a378a38df827dbd54896","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-08-18 11:06:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c95ef49cd5da189fb88dbb73394d3fa76de025b3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-18 11:14:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ad2b6d6cab93038ea667311046f0baf9f4d1238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:14:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f3c69b0b5d0d1d4dbe6c6b3d3f50c507dfbc2ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:59:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef22c8f8ef934fe3a38f880dbc72c9bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e8288f40474d2a9360b05726fb32b6 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ba3156267c140138babce3206f386ad : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e601c937a624947a13e79c30ff06e90 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4395495133c44effb7b4a1110858e805 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ceae2189c7d428fa622334d22ab4d26 : SUCCESS in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7371392fcb640619844fa8f6a64b349 : SUCCESS in 2m 50s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/30ee404977864723a0edd8cce23bc8e8 : SUCCESS in 35m 04s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/16f9bf19f9ad430f9b9d5f988b66051b : SUCCESS in 39m 34s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02677c398ad3435199273e83942b28a9 : SUCCESS in 36m 07s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/7d53df7ab1d24ae787485668292f1998 : SUCCESS in 21m 41s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/39d8dff7e1bb4936b17aeb1be53c21b6 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"955a3e83340c4328d2adc6378e24ae4f3122bdfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:59:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ece958bac27c7180e809cf7b6f3c14e02f1aef8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 12:00:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee1d05867bb4d1ea0aa6463d9fd91c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b9f4cfcd88405eafe3ebb5220b7181 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27ba6f0313be44f4941c86b332800d8c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da8684aef168eeff48bece652482f8b7d0e457eb","revisions":{"55656048b977fdb2d4b40e597f964d7b24f85e31":{"kind":"REWORK","_number":1,"created":"2026-07-24 14:06:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/998624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/24/998624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/998624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/998624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/998624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/24/998624/1"}}},"commit":{"parents":[{"commit":"42b608c4744f5cfa7ebe93bcbb5dda836249c2b2","subject":"Add distributed coordination for freezer-scheduler clustering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/42b608c4744f5cfa7ebe93bcbb5dda836249c2b2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 14:04:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 14:06:00.000000000","tz":120},"subject":"Avoid rotation of manually created cinder-backups","message":"Avoid rotation of manually created cinder-backups\n\nExisting rotation logic for cinder-backup does not account for\nmanually created backups. This means, that if user has created a backup\nfor the same volume, that is managed by freezer - this backup\nwill be rotated by freezer.\n\nIn order to fix the issue, we leverage new /backups PATH method,\nwhere backup record is being created, so we have backup UUID.\nThis uuid is written down to cinder-backup metadata, while\ncinder-backup UUID is pushed to the /backups through PATCH, as well as\nthe status update.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/998519\nChange-Id: I30f2c8ec6444c9e00a33c770eb269f53fd41d114\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/55656048b977fdb2d4b40e597f964d7b24f85e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/55656048b977fdb2d4b40e597f964d7b24f85e31"}]},"branch":"refs/heads/master"},"da8684aef168eeff48bece652482f8b7d0e457eb":{"kind":"REWORK","_number":2,"created":"2026-08-17 12:50:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/998624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/24/998624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/998624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/998624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/998624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/24/998624/2"}}},"commit":{"parents":[{"commit":"e738145344f8440d047e909d522a9d197a3cf196","subject":"scheduler: stop passing unused project_id to BackupsManager.create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e738145344f8440d047e909d522a9d197a3cf196"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 14:04:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-17 12:50:45.000000000","tz":120},"subject":"Avoid rotation of manually created cinder-backups","message":"Avoid rotation of manually created cinder-backups\n\nExisting rotation logic for cinder-backup does not account for\nmanually created backups. This means, that if user has created a backup\nfor the same volume, that is managed by freezer - this backup\nwill be rotated by freezer.\n\nIn order to fix the issue, we leverage new /backups PATH method,\nwhere backup record is being created, so we have backup UUID.\nThis uuid is written down to cinder-backup metadata, while\ncinder-backup UUID is pushed to the /backups through PATCH, as well as\nthe status update.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/998519\nChange-Id: I30f2c8ec6444c9e00a33c770eb269f53fd41d114\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/da8684aef168eeff48bece652482f8b7d0e457eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/da8684aef168eeff48bece652482f8b7d0e457eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
