)]}'
{"id":"openstack%2Ffuel-web~201217","triplet_id":"openstack%2Ffuel-web~master~I861485b5aebd8ae784fc71f405aab87e73f6d2e0","project":"openstack/fuel-web","branch":"master","topic":"bp/templates-for-networking","hashtags":[],"change_id":"I861485b5aebd8ae784fc71f405aab87e73f6d2e0","subject":"API for managing network groups","status":"MERGED","created":"2015-07-13 15:10:09.000000000","updated":"2016-04-19 06:44:49.000000000","submitted":"2015-07-23 14:16:41.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":80,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1118750606842745022bd08a8a9bca89c83222db","_number":201217,"virtual_id_number":201217,"owner":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2015-07-23 14:16:41.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-07-23 13:32:41.000000000","_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},{"value":0,"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},{"value":0,"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":1,"date":"2015-07-23 12:39:29.000000000","_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2015-07-23 14:22:58.000000000","post_submit":true,"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},{"value":0,"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},{"value":0,"_account_id":10959,"name":"Alexander Kislitsky","email":"akislitsky@mirantis.com","username":"akislitsky"},{"value":0,"_account_id":14543,"name":"Maciej Kwiek","email":"mkwiek@mirantis.com","username":"mkwiek"},{"value":0,"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},{"value":0,"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},{"value":0,"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},{"value":0,"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},{"value":0,"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},{"value":0,"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},{"value":0,"_account_id":15454,"name":"Sylwester Brzeczkowski","email":"sbrzeczkowski@mirantis.com","username":"sbrzeczkowski"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-07-23 14:15:21.000000000","_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},{"value":1,"date":"2015-07-23 14:06:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},{"value":1,"date":"2015-07-23 13:33:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},{"value":1,"date":"2015-07-23 12:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},{"value":0,"_account_id":10959,"name":"Alexander Kislitsky","email":"akislitsky@mirantis.com","username":"akislitsky"},{"value":0,"_account_id":14543,"name":"Maciej Kwiek","email":"mkwiek@mirantis.com","username":"mkwiek"},{"value":1,"date":"2015-07-23 12:49:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},{"value":1,"date":"2015-07-23 13:21:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},{"value":1,"date":"2015-07-23 14:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},{"value":1,"date":"2015-07-23 12:54:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},{"value":0,"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},{"value":0,"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},{"value":0,"_account_id":15454,"name":"Sylwester Brzeczkowski","email":"sbrzeczkowski@mirantis.com","username":"sbrzeczkowski"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-07-23 14:15:21.000000000","_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},{"value":0,"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},{"value":0,"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},{"value":0,"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},{"value":0,"_account_id":10959,"name":"Alexander Kislitsky","email":"akislitsky@mirantis.com","username":"akislitsky"},{"value":0,"_account_id":14543,"name":"Maciej Kwiek","email":"mkwiek@mirantis.com","username":"mkwiek"},{"value":0,"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},{"value":0,"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},{"value":0,"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},{"value":0,"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},{"value":0,"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},{"value":0,"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},{"value":0,"_account_id":15454,"name":"Sylwester Brzeczkowski","email":"sbrzeczkowski@mirantis.com","username":"sbrzeczkowski"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},{"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},{"_account_id":10959,"name":"Alexander Kislitsky","email":"akislitsky@mirantis.com","username":"akislitsky"},{"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},{"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},{"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},{"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},{"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},{"_account_id":14543,"name":"Maciej Kwiek","email":"mkwiek@mirantis.com","username":"mkwiek"},{"_account_id":15454,"name":"Sylwester Brzeczkowski","email":"sbrzeczkowski@mirantis.com","username":"sbrzeczkowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-07-23 07:58:30.000000000","updated_by":{"_account_id":15454,"name":"Sylwester Brzeczkowski","email":"sbrzeczkowski@mirantis.com","username":"sbrzeczkowski"},"reviewer":{"_account_id":15454,"name":"Sylwester Brzeczkowski","email":"sbrzeczkowski@mirantis.com","username":"sbrzeczkowski"},"state":"REVIEWER"},{"updated":"2015-07-23 07:59:26.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2015-07-23 08:07:46.000000000","updated_by":{"_account_id":10959,"name":"Alexander Kislitsky","email":"akislitsky@mirantis.com","username":"akislitsky"},"reviewer":{"_account_id":10959,"name":"Alexander Kislitsky","email":"akislitsky@mirantis.com","username":"akislitsky"},"state":"REVIEWER"},{"updated":"2015-07-23 08:08:58.000000000","updated_by":{"_account_id":14543,"name":"Maciej Kwiek","email":"mkwiek@mirantis.com","username":"mkwiek"},"reviewer":{"_account_id":14543,"name":"Maciej Kwiek","email":"mkwiek@mirantis.com","username":"mkwiek"},"state":"REVIEWER"},{"updated":"2015-07-23 09:58:16.000000000","updated_by":{"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},"reviewer":{"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},"state":"REVIEWER"},{"updated":"2015-07-23 12:39:29.000000000","updated_by":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2015-07-23 12:49:56.000000000","updated_by":{"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},"reviewer":{"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},"state":"REVIEWER"},{"updated":"2015-07-23 12:52:05.000000000","updated_by":{"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},"reviewer":{"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},"state":"REVIEWER"},{"updated":"2015-07-23 12:54:25.000000000","updated_by":{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},"reviewer":{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},"state":"REVIEWER"},{"updated":"2015-07-23 13:21:30.000000000","updated_by":{"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},"reviewer":{"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},"state":"REVIEWER"},{"updated":"2015-07-23 13:33:15.000000000","updated_by":{"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},"reviewer":{"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},"state":"REVIEWER"},{"updated":"2015-07-23 14:06:06.000000000","updated_by":{"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},"reviewer":{"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},"state":"REVIEWER"},{"updated":"2015-07-23 14:12:03.000000000","updated_by":{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},"reviewer":{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},"state":"REVIEWER"},{"updated":"2015-07-23 14:15:21.000000000","updated_by":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"reviewer":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"state":"REVIEWER"},{"updated":"2015-07-23 14:16:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-07-23 14:22:58.000000000","updated_by":{"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},"reviewer":{"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},"state":"REVIEWER"}],"messages":[{"id":"719c001d418eae34e203b5bcfcb705221d8fee59","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-13 15:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c4b73138c5c2a0a20f3eb6ba25ac0494cea4b03","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-13 15:10:17.000000000","message":"Patch Set 1:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2060/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"754b6b9424680d203932019532bed7c4d193b620","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-13 15:10:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/1/check/noop/0a9c0d1/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f121adba43e5ba19242be23ab7b2dfb000826751","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-13 15:10:19.000000000","message":"Patch Set 1:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2158/ (2/2)","accounts_in_message":[],"_revision_number":1},{"id":"1e86c38386fc490129280beafe150562bb553552","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-13 15:10:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0c450f343e7b56e268cc2acc637bb633e92fd376","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-13 15:21:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2060/ : SUCCESS\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2158/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d8adebeb903e4c9b3cc6b54ebf81f09acd49b315","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-20 15:31:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3fdb82bb974985c858db64997aa5460d7fb3fb7d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-20 15:31:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/2/check/noop/5df0aee/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e8513618c3e10ac175096f34dd78dd6e7281f718","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 15:31:37.000000000","message":"Patch Set 2:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2490/ (1/3)","accounts_in_message":[],"_revision_number":2},{"id":"d895cae60d791b3610e6fe65cf3f066174c9cc69","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 15:31:39.000000000","message":"Patch Set 2:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/216/ (2/3)","accounts_in_message":[],"_revision_number":2},{"id":"159a95ab271328ba9656dad3e9fe2a52cd227595","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 15:31:42.000000000","message":"Patch Set 2:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2595/ (3/3)","accounts_in_message":[],"_revision_number":2},{"id":"acf33101c6df77bf72b97b555e0b39de6930a821","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 15:42:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2490/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/216/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2595/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"54af900feeb1ea2828a0c24c66c91f8c3747a906","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-20 16:44:18.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a5773a944dd2ae0111df2eef1399405f8c0227f2","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-20 16:44:30.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"9d6ab16ceaf4141fe652c264608b30765c02075f","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-20 22:11:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bece2bff97b15bdbcc065e56afebe5dda8349fb9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-20 22:11:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/3/check/noop/b63af99/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"deddabd79b65c3937bc705e0ea031cc4b875d244","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 22:11:14.000000000","message":"Patch Set 3:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2513/ (1/3)","accounts_in_message":[],"_revision_number":3},{"id":"dec1b36d27017e3ba55c133372e03c27f6faf324","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 22:11:16.000000000","message":"Patch Set 3:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/239/ (2/3)","accounts_in_message":[],"_revision_number":3},{"id":"bfb3b2f37ae0e5971bb0757ce27982c172232526","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 22:11:19.000000000","message":"Patch Set 3:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2620/ (3/3)","accounts_in_message":[],"_revision_number":3},{"id":"7103a220730fed6e604f2d6472920baf55ab54db","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-20 22:11:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"391a33c1f5665e3018945bedf947035f1830b613","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-20 22:16:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2513/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/239/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2620/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2c10632a07a083b81f270a1097629b4ef99a716c","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-21 22:21:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f4368fdcdb5b58909fb73027ba512650743cce80","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:21:52.000000000","message":"Patch Set 4:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2591/ (1/3)","accounts_in_message":[],"_revision_number":4},{"id":"bea991cceab1e46781676d7d2411e866ede347c8","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:21:54.000000000","message":"Patch Set 4:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2698/ (2/3)","accounts_in_message":[],"_revision_number":4},{"id":"5d8657ca6255d3800dcbe8bce5115a66337aa78d","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:21:56.000000000","message":"Patch Set 4:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/317/ (3/3)","accounts_in_message":[],"_revision_number":4},{"id":"5a3d1c6c84697e5258a7c70099b6afbbfcacce39","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:31:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2591/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/317/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2698/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"9c285deda8a87edcf3d96d8fe97020be3f6e7bb6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-21 22:34:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/4/check/noop/e8927f7/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"026dd176810f16a09a8559ab488790c7138494cf","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-21 22:50:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a29c0969104fd8e518fb11ba94cf5edcada37ff6","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:50:59.000000000","message":"Patch Set 5:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2592/ (1/3)","accounts_in_message":[],"_revision_number":5},{"id":"11653484555d041369d492486a853557d2e14807","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:51:01.000000000","message":"Patch Set 5:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/318/ (2/3)","accounts_in_message":[],"_revision_number":5},{"id":"fd54081eeb155db5920f73a281aee75488d3e6af","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 22:51:03.000000000","message":"Patch Set 5:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2699/ (3/3)","accounts_in_message":[],"_revision_number":5},{"id":"daf948a2481c186e759602e8962f5bcb2dfb630e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-21 22:54:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/5/check/noop/9dcd417/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"01a63968e107b1fbe5f94403d4dd644436821bef","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 23:02:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2592/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/318/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2699/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"0cef6ad43aecb47a623b596ef92f613b98f26e5f","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-21 23:18:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"be3a7c6669680158fbb6ae7c568fb8a04f5745c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-21 23:18:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/6/check/noop/0329b72/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"1328963ef8d85da070b204129c2dd3a54b5b21fb","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 23:18:41.000000000","message":"Patch Set 6:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2593/ (1/3)","accounts_in_message":[],"_revision_number":6},{"id":"f140b977b2a17e8c1a05290a95f0f461b98bf20a","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 23:18:44.000000000","message":"Patch Set 6:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/319/ (2/3)","accounts_in_message":[],"_revision_number":6},{"id":"d4f0d3b6cbc8eaf167c844385e70fc7ea1537c08","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 23:18:46.000000000","message":"Patch Set 6:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2700/ (3/3)","accounts_in_message":[],"_revision_number":6},{"id":"615853deb05fc83c4ea639e167f02bc6fa7c4f70","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-21 23:30:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild Successful \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2593/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/319/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2700/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"98f1fe2dc7953d261d14ff99d7d824ccc44aaf2d","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-22 14:46:40.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"938360b3d37d7dd3b4e87076c7f62082aa748047","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-22 14:48:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f32db5fba2768ec10b88e8697b4bf0066e2765c2","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-22 15:00:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"30d634c2d274e26f5545504ccb4b48de4324ed23","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-22 15:31:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"70a56a952b86c6c7af6981d94e219bde274783be","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-22 15:58:57.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"01107eeca7d4b0ba5f2bb1a8ca67a6cfd6b35ebf","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-22 16:03:08.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d3b0c205c80f64f5f2ab70b6098fa73f69c54271","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-22 16:14:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3f3db9e65653438a552d5ccd105a0a65576dfc12","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-22 22:20:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"970ff23a2e2cee3ff740097d3c03ccfa2ebbc571","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-22 22:20:36.000000000","message":"Patch Set 7:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/386/ (1/3)","accounts_in_message":[],"_revision_number":7},{"id":"76d02713b673cb48c9a4a5c4cd1dbc020391c734","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-22 22:20:38.000000000","message":"Patch Set 7:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2663/ (2/3)","accounts_in_message":[],"_revision_number":7},{"id":"7256cf1240e268f8c0407a7b401547fcd2b352b8","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-22 22:20:40.000000000","message":"Patch Set 7:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2774/ (3/3)","accounts_in_message":[],"_revision_number":7},{"id":"3f09478440e8e8bc93c4ec8aee22ae11e89ee0dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-22 22:21:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/7/check/noop/0d8bb03/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"0ba946f99af21157392eff6b3decb2628f27dca6","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-22 22:33:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2663/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/386/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2774/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"ddf511d31c1a4389aa53db105fa6d4dc946829e8","author":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"date":"2015-07-23 04:07:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a451795d70ded813b5db93c0296cf62994b77e02","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 04:08:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/8/check/noop/745f711/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"0db3726e5521c06ed152a114a6450d0f94d2998c","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 04:08:10.000000000","message":"Patch Set 8:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2668/ (1/3)","accounts_in_message":[],"_revision_number":8},{"id":"f2d1a8f977eb34b22675e6ead27e31c569e70da0","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 04:08:12.000000000","message":"Patch Set 8:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/391/ (2/3)","accounts_in_message":[],"_revision_number":8},{"id":"0095f784eda5633a3ebd1c2d098e981ee9f8844c","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 04:08:15.000000000","message":"Patch Set 8:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2778/ (3/3)","accounts_in_message":[],"_revision_number":8},{"id":"de13d302d7077837df5b3b6543a2c74d37903fdc","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 04:21:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild Successful \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2668/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/391/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2778/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"f7609ca2a815267f2013bda9c4c3980c3397b529","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 07:46:00.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"45ed8bed3a9eaac77fcdf601e7e5353c767f1270","author":{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},"date":"2015-07-23 07:54:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1594b19a793ccc661be2a3b492088ac2be10a851","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 07:56:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"407107923cd4232d4931407cc742c0867cb11f59","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 07:57:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"68edcd8304dbc718cc41f40407865be627ac1876","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 07:58:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/9/check/noop/240bc12/ : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"f7b71ffb08c5462065a0cd2a3f169065d17abdb2","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 07:58:06.000000000","message":"Patch Set 9:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2669/ (1/3)","accounts_in_message":[],"_revision_number":9},{"id":"f35da60eaa75781cc92783178d615c0691c2489b","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 07:58:11.000000000","message":"Patch Set 9:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2779/ (2/3)","accounts_in_message":[],"_revision_number":9},{"id":"b8d95b86c7f9e3308d0d5e43fd5523406f87f26f","author":{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},"date":"2015-07-23 07:58:41.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"d28d00f1914d5e2b9d06310c181f93fd8fdac1db","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 08:08:54.000000000","message":"Patch Set 9:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/392/ (3/3)","accounts_in_message":[],"_revision_number":9},{"id":"8c4600e38bb5cd50b36fcbdaf16d0d63371c72fb","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 08:11:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2669/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/392/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2779/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"10f12b87c491cdac18d547d7d55949a28a7c9aa0","author":{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},"date":"2015-07-23 08:18:03.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0581f34d25623c78bd9ddbd7a7a3e6f29f321445","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 08:24:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3683a662df464011fd043c1a7d2ba0e59886024a","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 08:57:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e11148e0178119a7976f595b18184e82f89e9528","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 08:57:44.000000000","message":"Patch Set 10:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2670/ (1/3)","accounts_in_message":[],"_revision_number":10},{"id":"70194aadedce3fa9dd70a39206b7b0b64f6f1b1b","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 08:57:45.000000000","message":"Patch Set 10:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/393/ (2/3)","accounts_in_message":[],"_revision_number":10},{"id":"3581234f98a50ef5acab671025a86ffb3a713d46","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 08:57:49.000000000","message":"Patch Set 10:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2780/ (3/3)","accounts_in_message":[],"_revision_number":10},{"id":"97876ec30e2bb2bde3fd6d97d42717988fcf0842","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 08:58:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/10/check/noop/56a6824/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"71ed343ee0dc3091961fbf578f5de0b36c6b5c2e","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 09:10:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild Successful \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2670/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/393/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2780/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"4f312931c5ac35b17fe76e04c1fc65399ae4e21d","author":{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},"date":"2015-07-23 09:16:33.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"05c6d086d26abf0d875941e07f59595caf002d24","author":{"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},"date":"2015-07-23 09:34:05.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"b87a0a2440946410bc37e5d7ee0a0762ae35425a","author":{"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},"date":"2015-07-23 09:38:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4d81f57036b11c96e423d07b7b38c53aaf0f7e49","author":{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},"date":"2015-07-23 09:40:20.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"d579e02385cdb26f5c51640a390dbc4dfc1a18d8","author":{"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},"date":"2015-07-23 09:50:47.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2fc2dd311ea80d8a31932f27e01515df7476c85c","author":{"_account_id":11898,"name":"Alexander Saprykin","email":"cutwatercore@gmail.com","username":"asaprykin"},"date":"2015-07-23 09:58:16.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4bed97fb5136f33a0b97d83780caadbe9d7384b2","author":{"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},"date":"2015-07-23 10:07:35.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b1823b2df2f8aa12022268c04f1c99557d819636","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 10:18:42.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7ecaea6c39254a05663d99f2bc319c9492e8a80a","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 11:26:20.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b16f87063624816c3a80767a3cf0b78a67635ccf","author":{"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},"date":"2015-07-23 11:57:11.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8455a7d164384212cc8d9d52bdea42923bf8f69c","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 12:16:53.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1b76fd8875e3c6270dbe536fb1653b7d54a34fa7","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 12:17:43.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"227247d119ed3edd9d3e191552d0272f21c99432","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:17:51.000000000","message":"Patch Set 11:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2687/ (1/3)","accounts_in_message":[],"_revision_number":11},{"id":"0b66d45b467f33f29bc602a84f078dbec1463213","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 12:17:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/11/check/noop/57f455f/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"b2e2c915236efb8371748215b90854c965150a48","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:17:53.000000000","message":"Patch Set 11:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2798/ (2/3)","accounts_in_message":[],"_revision_number":11},{"id":"cc8a66e095d18db54344da773ef7539098b72453","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:17:55.000000000","message":"Patch Set 11:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/410/ (3/3)","accounts_in_message":[],"_revision_number":11},{"id":"a26fb5d8db97d3babede8da05fce1d820f56d75b","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:23:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild Failed \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2687/ : FAILURE\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/410/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2798/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"c2c2e2bd0ebb4801f507ad06e3849497143109df","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 12:25:32.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ccc9fba71549a10cf26b4f2284e605fc26b07b34","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 12:25:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/17/201217/12/check/noop/3ee6dea/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"7c289aea813c878a79dce04fce078357bf58c4fd","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:25:43.000000000","message":"Patch Set 12:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web-docs/2689/ (1/3)","accounts_in_message":[],"_revision_number":12},{"id":"16826f6563f59a430700fbe6909e6285d207ffa0","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:25:45.000000000","message":"Patch Set 12:\n\nBuild Started https://ci.fuel-infra.org/job/fuel-web-linkcheck/412/ (2/3)","accounts_in_message":[],"_revision_number":12},{"id":"f45d7a6c6d78ebe58282752811ec79e418669375","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:25:47.000000000","message":"Patch Set 12:\n\nBuild Started https://ci.fuel-infra.org/job/verify-fuel-web/2800/ (3/3)","accounts_in_message":[],"_revision_number":12},{"id":"6e56de89a8ce2e132ae8a524fa1cccb52efcebca","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 12:39:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild Successful \n\n* verify-fuel-web-docs https://ci.fuel-infra.org/job/verify-fuel-web-docs/2689/ : SUCCESS\n\n* fuel-web-linkcheck https://ci.fuel-infra.org/job/fuel-web-linkcheck/412/ : SUCCESS (skipped)\n\n* verify-fuel-web https://ci.fuel-infra.org/job/verify-fuel-web/2800/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"ac575f6cf556b6946d8ad92d1b98f2c28c5a9110","author":{"_account_id":14167,"name":"Andriy Popovych","email":"apopovych@mirantis.com","username":"apopovych"},"date":"2015-07-23 12:49:56.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"44cf12e46cc70a9e7f34840b8955df07474ce792","author":{"_account_id":12559,"name":"Valyavskiy Viacheslav","email":"vvalyavskiy@mirantis.com","username":"vvalyavskiy"},"date":"2015-07-23 12:52:05.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"9589e242e46321bdba4f5cf74c90a676acd6445e","author":{"_account_id":11577,"name":"Ivan Kliuk","email":"ivan.kliuk+spam@gmail.com","username":"ivan"},"date":"2015-07-23 12:54:25.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"0dc52974e0b7f3666b5b62f2e6a4cc9ae0a6aced","author":{"_account_id":10391,"name":"Ihor Kalnytskyi","email":"ihor@kalnytskyi.com","username":"ikalnitsky"},"date":"2015-07-23 13:21:30.000000000","message":"Patch Set 12: Code-Review+1\n\n(4 comments)\n\nI spent few minor comments. I\u0027m ok if you resolve them in separate patch if you wish.","accounts_in_message":[],"_revision_number":12},{"id":"99bc31d38d9560e90acdbb12b81a179cdb7537c4","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 13:32:41.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"3a35bbb92de408f4aa00355a73fa1752c5a52a45","author":{"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},"date":"2015-07-23 13:33:15.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"713e776034c25d2974055048493931a994f2376c","author":{"_account_id":11081,"name":"Artem Panchenko","email":"artem@panchenko.email","username":"apanchenko"},"date":"2015-07-23 13:34:31.000000000","message":"Patch Set 12:\n\npassed BVT http://jenkins-product.srt.mirantis.net:8080/view/custom_iso/job/7.0.custom.ubuntu.bvt_2/140/","accounts_in_message":[],"_revision_number":12},{"id":"108dd3c03b232d3ea6be6de4a401450c830a8ee7","author":{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},"date":"2015-07-23 13:58:54.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1b2c8e985e38c44f40e46b8937dbb3df5baa16df","author":{"_account_id":12200,"name":"Sebastian Kalinowski","email":"sebastian@kalinowski.eu","username":"prmtl"},"date":"2015-07-23 14:06:06.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b21c2c7d973c3a2d81cab2983061743dd8710e4d","author":{"_account_id":10488,"name":"Artur Svechnikov","email":"asvechnikov@mirantis.com","username":"asvechnikov"},"date":"2015-07-23 14:12:03.000000000","message":"Patch Set 12: Code-Review+1\n\ncomments will be fixed as a bug","accounts_in_message":[],"_revision_number":12},{"id":"f873c8cb28204d4b2fbcc0ca7f00cdf91650845f","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 14:13:48.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"bc0bd170e2e0bb7377e7fb2d5730a29f1d38572f","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 14:15:21.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"eb865472b079a57eb1a1822df422a490b447831b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 14:16:28.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"d898793b07cb561a28b288ae332a3f5151a0f5d7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 14:16:41.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop http://logs.openstack.org/17/201217/12/gate/noop/87fafe6/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"1751b987533710e30630792ccd8bdadafc826d9d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-23 14:16:43.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":12},{"id":"688c721a808279d080abc3796cb9b1b8a9c828a3","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 14:16:57.000000000","message":"Patch Set 12:\n\nBuild Started https://ci.fuel-infra.org/job/gate-fuel-web/427/","accounts_in_message":[],"_revision_number":12},{"id":"40564ce7a2f423d0f03e54da8f1b6795a22ecdf9","author":{"_account_id":10443,"name":"Fedor Zhadaev","email":"fzhadaev@mirantis.com","username":"fzhadaev"},"date":"2015-07-23 14:22:58.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"44a9511ccbdc0d5695f5d94db139e5dcc4198cb9","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2015-07-23 14:30:00.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* gate-fuel-web https://ci.fuel-infra.org/job/gate-fuel-web/427/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"d3f50a7bf121fed20a1aa0a899a043fe1e149fbc","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2015-07-23 15:00:44.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"84be725fad787f0fbcb5b4c370de188244304d5b","author":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"date":"2016-04-19 06:44:49.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"7084177cae802695277f0c59e04e6f2fec4e0760","revisions":{"8a6f3cfe125542979272fc312e9b004dc1233545":{"kind":"REWORK","_number":1,"created":"2015-07-13 15:10:09.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/1"}}},"commit":{"parents":[{"commit":"ec51a0dfddb782e579308a7e206bd10837562070","subject":"Merge \"Add restrictions for admin bond\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/ec51a0dfddb782e579308a7e206bd10837562070"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 15:08:54.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/8a6f3cfe125542979272fc312e9b004dc1233545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/8a6f3cfe125542979272fc312e9b004dc1233545"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ec51a0dfddb782e579308a7e206bd10837562070","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a07c5dd768b5997cc436049202d58d3be30a2592":{"kind":"REWORK","_number":2,"created":"2015-07-20 15:31:26.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/2"}}},"commit":{"parents":[{"commit":"f9dc26188bf40c408233187474599f01260388a6","subject":"[TBS3] Notification badge animation added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/f9dc26188bf40c408233187474599f01260388a6"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-20 15:29:15.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/a07c5dd768b5997cc436049202d58d3be30a2592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/a07c5dd768b5997cc436049202d58d3be30a2592"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9dc26188bf40c408233187474599f01260388a6","is_merged_in_target_branch":true,"change_id":"I2146afae30d02e7be774f89f4fba58caa6bd4af6","change_number":180118,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fa2287aaa82a0493f78076bb668044257370795e":{"kind":"REWORK","_number":3,"created":"2015-07-20 22:11:04.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/3"}}},"commit":{"parents":[{"commit":"f9dc26188bf40c408233187474599f01260388a6","subject":"[TBS3] Notification badge animation added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/f9dc26188bf40c408233187474599f01260388a6"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-20 22:10:27.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/fa2287aaa82a0493f78076bb668044257370795e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/fa2287aaa82a0493f78076bb668044257370795e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9dc26188bf40c408233187474599f01260388a6","is_merged_in_target_branch":true,"change_id":"I2146afae30d02e7be774f89f4fba58caa6bd4af6","change_number":180118,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9bed2deb568da640c441ebc4dac3f1fa65da6940":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2015-07-21 22:21:43.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/4"}}},"commit":{"parents":[{"commit":"01368e907b34180268841a97657cc5e7e4db843f","subject":"Merge \"Fixes for node label management in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/01368e907b34180268841a97657cc5e7e4db843f"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-21 22:21:43.000000000","tz":0},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/9bed2deb568da640c441ebc4dac3f1fa65da6940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/9bed2deb568da640c441ebc4dac3f1fa65da6940"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"01368e907b34180268841a97657cc5e7e4db843f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1b9fb579978a6b402c9b7c3db14b33caaecbd5ae":{"kind":"REWORK","_number":5,"created":"2015-07-21 22:50:49.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/5"}}},"commit":{"parents":[{"commit":"01368e907b34180268841a97657cc5e7e4db843f","subject":"Merge \"Fixes for node label management in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/01368e907b34180268841a97657cc5e7e4db843f"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-21 22:47:58.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/1b9fb579978a6b402c9b7c3db14b33caaecbd5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/1b9fb579978a6b402c9b7c3db14b33caaecbd5ae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"01368e907b34180268841a97657cc5e7e4db843f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"15b73d97daf168321d338cac2768a4c6910f47e5":{"kind":"REWORK","_number":6,"created":"2015-07-21 23:18:31.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/6"}}},"commit":{"parents":[{"commit":"01368e907b34180268841a97657cc5e7e4db843f","subject":"Merge \"Fixes for node label management in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/01368e907b34180268841a97657cc5e7e4db843f"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-21 23:18:23.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/15b73d97daf168321d338cac2768a4c6910f47e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/15b73d97daf168321d338cac2768a4c6910f47e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"01368e907b34180268841a97657cc5e7e4db843f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"44bd899fdb0914fe49c6a562e96cca1491240965":{"kind":"REWORK","_number":7,"created":"2015-07-22 22:20:26.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/7"}}},"commit":{"parents":[{"commit":"52a413e672b07396245bad16335870fffd8b7546","subject":"Merge \"[Nailgun] Get network group from networking template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/52a413e672b07396245bad16335870fffd8b7546"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-22 22:20:06.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/44bd899fdb0914fe49c6a562e96cca1491240965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/44bd899fdb0914fe49c6a562e96cca1491240965"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a413e672b07396245bad16335870fffd8b7546","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0455ed0113c803c359f5faac2a8e93431078c635":{"kind":"REWORK","_number":8,"created":"2015-07-23 04:07:59.000000000","uploader":{"_account_id":8829,"name":"Ryan Moe","email":"rmoe@mirantis.com","username":"rmoe"},"ref":"refs/changes/17/201217/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/8"}}},"commit":{"parents":[{"commit":"52a413e672b07396245bad16335870fffd8b7546","subject":"Merge \"[Nailgun] Get network group from networking template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/52a413e672b07396245bad16335870fffd8b7546"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-23 04:06:17.000000000","tz":-420},"subject":"API for managing network groups","message":"API for managing network groups\n\nIndividual network groups can be created, deleted and\nmodified. IP range management for each network is still\nhandled by the standard network_configuration API handlers.\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/0455ed0113c803c359f5faac2a8e93431078c635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/0455ed0113c803c359f5faac2a8e93431078c635"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a413e672b07396245bad16335870fffd8b7546","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3295b5c63e3b4f08ae6d4d5f3ea3c6c86d905645":{"kind":"REWORK","_number":9,"created":"2015-07-23 07:57:57.000000000","uploader":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"ref":"refs/changes/17/201217/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/9"}}},"commit":{"parents":[{"commit":"52a413e672b07396245bad16335870fffd8b7546","subject":"Merge \"[Nailgun] Get network group from networking template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/52a413e672b07396245bad16335870fffd8b7546"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","date":"2015-07-23 07:57:26.000000000","tz":180},"subject":"API for managing network groups","message":"API for managing network groups\n\nIndividual network groups can be created, deleted and\nmodified. IP range management for each network is still\nhandled by the standard network_configuration API handlers.\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/3295b5c63e3b4f08ae6d4d5f3ea3c6c86d905645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/3295b5c63e3b4f08ae6d4d5f3ea3c6c86d905645"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a413e672b07396245bad16335870fffd8b7546","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"201be8f7b76a34d992f521e4d1fac79393ce4e19":{"kind":"REWORK","_number":10,"created":"2015-07-23 08:57:34.000000000","uploader":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"ref":"refs/changes/17/201217/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/10"}}},"commit":{"parents":[{"commit":"52a413e672b07396245bad16335870fffd8b7546","subject":"Merge \"[Nailgun] Get network group from networking template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/52a413e672b07396245bad16335870fffd8b7546"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","date":"2015-07-23 08:57:19.000000000","tz":180},"subject":"API for managing network groups","message":"API for managing network groups\n\nIndividual network groups can be created, deleted and\nmodified. IP range management for each network is still\nhandled by the standard network_configuration API handlers.\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/201be8f7b76a34d992f521e4d1fac79393ce4e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/201be8f7b76a34d992f521e4d1fac79393ce4e19"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a413e672b07396245bad16335870fffd8b7546","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"68b018284ff1833e8a9e60590e6b8be149e7d8d4":{"kind":"REWORK","_number":11,"created":"2015-07-23 12:17:43.000000000","uploader":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"ref":"refs/changes/17/201217/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/11"}}},"commit":{"parents":[{"commit":"52a413e672b07396245bad16335870fffd8b7546","subject":"Merge \"[Nailgun] Get network group from networking template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/52a413e672b07396245bad16335870fffd8b7546"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","date":"2015-07-23 12:17:25.000000000","tz":180},"subject":"API for managing network groups","message":"API for managing network groups\n\nIndividual network groups can be created, deleted and\nmodified. IP range management for each network is still\nhandled by the standard network_configuration API handlers.\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/68b018284ff1833e8a9e60590e6b8be149e7d8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/68b018284ff1833e8a9e60590e6b8be149e7d8d4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a413e672b07396245bad16335870fffd8b7546","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7084177cae802695277f0c59e04e6f2fec4e0760":{"kind":"REWORK","_number":12,"created":"2015-07-23 12:25:32.000000000","uploader":{"_account_id":8392,"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","username":"aleksey"},"ref":"refs/changes/17/201217/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/fuel-web","ref":"refs/changes/17/201217/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/fuel-web refs/changes/17/201217/12"}}},"commit":{"parents":[{"commit":"52a413e672b07396245bad16335870fffd8b7546","subject":"Merge \"[Nailgun] Get network group from networking template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/52a413e672b07396245bad16335870fffd8b7546"}]}],"author":{"name":"Ryan Moe","email":"rmoe@mirantis.com","date":"2015-07-13 14:43:34.000000000","tz":-420},"committer":{"name":"Aleksey Kasatkin","email":"akasatkin@mirantis.com","date":"2015-07-23 12:25:22.000000000","tz":180},"subject":"API for managing network groups","message":"API for managing network groups\n\nIndividual network groups can be created, deleted and\nmodified. IP range management for each network is still\nhandled by the standard network_configuration API handlers.\n\nChange-Id: I861485b5aebd8ae784fc71f405aab87e73f6d2e0\nImplements: blueprint templates-for-networking\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/7084177cae802695277f0c59e04e6f2fec4e0760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/fuel-web/commit/7084177cae802695277f0c59e04e6f2fec4e0760"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"52a413e672b07396245bad16335870fffd8b7546","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
