)]}'
{"id":"openstack%2Ffuturist~975083","triplet_id":"openstack%2Ffuturist~master~I7a13d49db7f2ee93293dff420275f6cd34a5b569","project":"openstack/futurist","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-09 12:56:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7a13d49db7f2ee93293dff420275f6cd34a5b569","subject":"trivial: Remove redundant args, kwargs","status":"MERGED","created":"2026-01-28 18:35:50.000000000","updated":"2026-03-09 14:46:36.000000000","submitted":"2026-03-09 14:45:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975083-typing","meta_rev_id":"1fa87425345cfd8674bce9c5a35796e02ee10709","_number":975083,"virtual_id_number":975083,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 14:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-09 12:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-09 12:56:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 18:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8296221430fcf7716b6a4ce45edd81347291b354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-28 18:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ed0534c5ad1d8c43fcb25e3b44952fbf28c27d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 18:59:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/468286e27a5945c58fa3ac0333433879\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86d771fac6de40ad99f7a995e2e1f2ab : FAILURE in 21m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6210409630e54f99960034c298dcd9cd : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54e0eaccdbd1406eb6e308bde36e6dc6 : FAILURE in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76eae3fd444e4e9ba57de2c0e1cb76c7 : FAILURE in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd22b54ca33649c0a1eae77a3ecce205 : FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f99c955e034de55bda420b2849fd166c17ca47fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-29 12:44:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"28ff2c5058e51e8e128d3dc477830f62b4c05db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 14:31:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54d2a99a22894285a8fa15ff6898aa3b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c642fd5d5f349408f4ba3eb09727738 : SUCCESS in 1h 45m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f46efdeced14c9885e19dc1d8c32f17 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ee5c7ae9ffe45799d6bfa0587243d96 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d75e9837e0d46658753cd60b9a733f9 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b05fc7256336483086f35fb22db57be8 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"aab45c80c477fb79caad0811604a1f45ea0dd4a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 12:56:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"25bddffba824d02f8e367f0c5bbbfb2bc5b6651a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 12:56:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2f8565f5a76f8bd01444c440822cc5edd5494cf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:45:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d18354a0a674e09a88879bd1b04abb9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fec5751d84ff4915991939202778e99a : SUCCESS in 1h 44m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7f1abc478f4261803a5236de7b5d3a : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d44fd58ba20349fbb1e264b68a5d4ffa : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b48053fd4fa842b38ab7c9f37dc37fd8 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c734b393918b4c13b03063f6091e56bd : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"190a4ebb8d79536978c2e3a2c97780c49227165d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:45:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1fa87425345cfd8674bce9c5a35796e02ee10709","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:46:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b69109ef9b274dfba947023e079d9534\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb324e3765349339793e49195acd39a : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8660ca02c162b5af649e8eeedc0cfbb9ca81692","revisions":{"82a62a1b3b7b27c57ee7873db8625a6291cd0562":{"kind":"REWORK","_number":1,"created":"2026-01-28 18:35:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/975083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/futurist","ref":"refs/changes/83/975083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/futurist refs/changes/83/975083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/futurist refs/changes/83/975083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/futurist refs/changes/83/975083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/futurist refs/changes/83/975083/1"}}},"commit":{"parents":[{"commit":"a458b199329ea13e2e4d6916ed4eeb21a4e71549","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/futurist/commit/a458b199329ea13e2e4d6916ed4eeb21a4e71549"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-16 15:51:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-28 18:18:26.000000000","tz":0},"subject":"trivial: Remove redundant args, kwargs","message":"trivial: Remove redundant args, kwargs\n\nMost of these threading classes do not accept any arguments, while\nthreading.Condition accepts a mere one. We can therefore simplify things\ndramatically.\n\nNote that we \"lie\" in our type hints for GreenThreading. That\u0027s okay\nsince these things should behave identically for our purposes.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7a13d49db7f2ee93293dff420275f6cd34a5b569\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/futurist/commit/82a62a1b3b7b27c57ee7873db8625a6291cd0562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/futurist/commit/82a62a1b3b7b27c57ee7873db8625a6291cd0562"}]},"branch":"refs/heads/master"},"a8660ca02c162b5af649e8eeedc0cfbb9ca81692":{"kind":"REWORK","_number":2,"created":"2026-01-29 12:44:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/975083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/futurist","ref":"refs/changes/83/975083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/futurist refs/changes/83/975083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/futurist refs/changes/83/975083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/futurist refs/changes/83/975083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/futurist refs/changes/83/975083/2"}}},"commit":{"parents":[{"commit":"a458b199329ea13e2e4d6916ed4eeb21a4e71549","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/futurist/commit/a458b199329ea13e2e4d6916ed4eeb21a4e71549"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-16 15:51:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-29 12:38:04.000000000","tz":0},"subject":"trivial: Remove redundant args, kwargs","message":"trivial: Remove redundant args, kwargs\n\nMost of these threading classes do not accept any arguments, while\nthreading.Condition accepts a mere one. We can therefore simplify things\ndramatically.\n\nNote that we \"lie\" in our type hints for GreenThreading. That\u0027s okay\nsince these things should behave identically for our purposes.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7a13d49db7f2ee93293dff420275f6cd34a5b569\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/futurist/commit/a8660ca02c162b5af649e8eeedc0cfbb9ca81692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/futurist/commit/a8660ca02c162b5af649e8eeedc0cfbb9ca81692"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
