)]}'
{"id":"openstack%2Fglance-specs~656895","triplet_id":"openstack%2Fglance-specs~master~I92940d839a38fe848dfa586a9bc2642848078eab","project":"openstack/glance-specs","branch":"master","topic":"barbican-secret-deletion-support","hashtags":[],"change_id":"I92940d839a38fe848dfa586a9bc2642848078eab","subject":"Add spec for barbican secret deletion","status":"MERGED","created":"2019-05-02 22:43:26.000000000","updated":"2019-08-21 14:48:01.000000000","submitted":"2019-08-21 14:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656895-1566398881319-16970a7d","meta_rev_id":"11a2707f3e90f5aeb169aafbf683bfd987705ea4","_number":656895,"virtual_id_number":656895,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-23 19:38:18.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":2,"date":"2019-08-21 14:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-21 14:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-08-21 14:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2019-07-25 05:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-08-21 14:37:18.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 17:16:00.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2019-06-03 15:30:24.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2019-06-11 05:46:19.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-25 05:57:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-08-21 14:33:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-21 14:37:18.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-08-21 14:48:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c29bbc2c11bc068af1703427cf349f0f66037ac","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-02 22:43:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"818ea3e79b7954f2c835250a3c44a5947ea87bdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:53:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/656895/1/check/openstack-tox-docs/b48dfaa/html/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c39cb00a3396a2fabaac81738d0a0cb6a53cfdf2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2019-05-20 17:16:00.000000000","message":"Patch Set 1:\n\nThere is some intersection with\n    https://review.opendev.org/#/c/609667/\n\nWhich needs to be thought through if that spec is going to be implemented.","accounts_in_message":[],"_revision_number":1},{"id":"f2e0b4aef3bb00a2984dec5dbe997e4ab5e5d67b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-21 06:32:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1fddf29af7edc2e4ce99e8499f3f8fc56a8653ca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-03 15:15:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3caac84d2a6660dc6f40dd76a05ee85389c5eaf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-03 15:17:13.000000000","message":"Patch Set 2:\n\nNew revision in light of discussion at the weekly glance meeting on 23 May: http://eavesdrop.openstack.org/meetings/glance/2019/glance.2019-05-23-14.00.log.html#l-146","accounts_in_message":[],"_revision_number":2},{"id":"36b940d37c31f8dfbfedeadcc2e26e567daadd25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:38:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/656895/2/check/openstack-tox-docs/5a8535b/ : RETRY_LIMIT in 1m 30s","accounts_in_message":[],"_revision_number":2},{"id":"dfc6ef93843a8a01fbf6a00f80ec765ad4e8307a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-03 16:58:12.000000000","message":"Patch Set 2:\n\nLooks similar to https://bugs.launchpad.net/openstack-gate/+bug/1831474 except that the problem is happening during update apt cache instead of during pip install.","accounts_in_message":[],"_revision_number":2},{"id":"06f34447e7d30beed8b1ea89080eee976e77bae7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-03 16:58:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"de010b1dab41b04afb2fd5ce5b31f98e7e3dc689","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 17:08:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/656895/2/check/openstack-tox-docs/6e74d2d/html/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e5cf940e76dd49259597ec5c4484a8afd4f6d084","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 08:25:00.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nHi Brian,\n\nWhile writing PoC I came across below findings, please suggest.\n\nThank you!!","accounts_in_message":[],"_revision_number":2},{"id":"2f880c982d02c2bede158d6ce861fde4f454a4bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 09:07:11.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nminor spelling correction","accounts_in_message":[],"_revision_number":2},{"id":"93c2e2b64a602ca0c6c020ab514f0f2b35fb0b3b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-07 09:56:16.000000000","message":"Patch Set 2:\n\nPoC: https://review.opendev.org/#/c/663878/1","accounts_in_message":[],"_revision_number":2},{"id":"5a2cf04050698ba0eecbd8a5c68e90866909209a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-10 21:49:18.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n@Abhishek: Answer to your question inline.\n\nThanks for catching that typo, correction coming up shortly.","accounts_in_message":[],"_revision_number":2},{"id":"e99ace306a812a32ddc54c08ec48b011ed1f713d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-10 21:52:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a690e91bf6f8699f714e2b10eadcaf91909739b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:19:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/656895/3/check/openstack-tox-docs/13b7919/html/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2823e302d621d9db096407a500291bab4b0142ac","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-11 04:29:42.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good, Thank you Brian!!!","accounts_in_message":[],"_revision_number":3},{"id":"fbaff3d9639fb9faaf43370775b27435668b0b13","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-06-11 05:46:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"61a9c25de6d22e22afae60deace20bb989a354d2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-10 21:32:25.000000000","message":"Patch Set 3:\n\nJust want to mention the Barbican \u0027Consumer API\u0027 that\u0027s being introduced in Train: https://review.opendev.org/#/c/662013/\n\nI\u0027m not sure we can leverage it for this particular purpose, but maybe someone else will have some thoughts.","accounts_in_message":[],"_revision_number":3},{"id":"d356350d254f1b75269c15212d6e5563c0420dc4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-23 19:36:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c72fb903aea1be66e2a5a9175d36450620a33e73","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-23 19:38:18.000000000","message":"Patch Set 4:\n\nChange was remove the merge conflict caused by the index.html file, which is no longer needed by this patch.  No change to the spec.","accounts_in_message":[],"_revision_number":4},{"id":"69c2d04e8292ad2a05605c9b23ba331ddbc1ec2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 19:43:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/656895/4/check/openstack-tox-docs/f89fff0/html/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"2654edd112d20cd0d8da146e675a45807e5910b0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-25 05:57:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a11738822b3536009ca9d2d2725830c8e671e8e5","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-21 14:07:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ae0f55577a7221904ff15ea02be952f674d321a4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-21 14:33:42.000000000","message":"Patch Set 4: Code-Review+2\n\nMakes sense to me. Being explicit is good and makes it clear who is responsible for clean up.\n\n+1 to deprecating the old config option as well.","accounts_in_message":[],"_revision_number":4},{"id":"674bbe28ba9230f7d3f1d162579007f6d403bb3c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-21 14:37:18.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2875893b4efe59627b65046a4f258be4d1765988","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:37:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3f5031d9cae943d910f66b5586b69e6e13e0514d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:48:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb9c33a6b3e489ea7ee3c623adf9c15 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"47cc4e55e818479b639fe5770de48b371fdb30b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:48:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5e3cb4f599d584d90872d71f8b88e44620352f82","revisions":{"74ac19d8ec3ed8e7a1102fa92cc1db06cbe93467":{"kind":"REWORK","_number":1,"created":"2019-05-02 22:43:26.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/656895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/95/656895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/1"}}},"commit":{"parents":[{"commit":"4e095520a4093e803a24fb7e537435e568947ce8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/4e095520a4093e803a24fb7e537435e568947ce8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-02 22:25:47.000000000","tz":-360},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-02 22:43:26.000000000","tz":-360},"subject":"Add spec for barbican secret deletion","message":"Add spec for barbican secret deletion\n\nChange-Id: I92940d839a38fe848dfa586a9bc2642848078eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/74ac19d8ec3ed8e7a1102fa92cc1db06cbe93467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/74ac19d8ec3ed8e7a1102fa92cc1db06cbe93467"}]},"branch":"refs/heads/master"},"9bdbdfb0ec87adf2a203cc8d3ea45fe7f48937b4":{"kind":"REWORK","_number":2,"created":"2019-06-03 15:15:43.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/656895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/95/656895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/2"}}},"commit":{"parents":[{"commit":"d9fd388bda35d26fba721b3c217ef4f95ff21433","subject":"Move Stein implemented specs to \u0027implemented\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d9fd388bda35d26fba721b3c217ef4f95ff21433"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-02 22:25:47.000000000","tz":-360},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-06-03 15:15:31.000000000","tz":-240},"subject":"Add spec for barbican secret deletion","message":"Add spec for barbican secret deletion\n\nChange-Id: I92940d839a38fe848dfa586a9bc2642848078eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9bdbdfb0ec87adf2a203cc8d3ea45fe7f48937b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9bdbdfb0ec87adf2a203cc8d3ea45fe7f48937b4"}]},"branch":"refs/heads/master"},"fabcf02c0d376d663792656598d2e32b39fc89bc":{"kind":"REWORK","_number":3,"created":"2019-06-10 21:52:11.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/656895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/95/656895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/3"}}},"commit":{"parents":[{"commit":"d9fd388bda35d26fba721b3c217ef4f95ff21433","subject":"Move Stein implemented specs to \u0027implemented\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d9fd388bda35d26fba721b3c217ef4f95ff21433"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-02 22:25:47.000000000","tz":-360},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-06-10 21:51:59.000000000","tz":-240},"subject":"Add spec for barbican secret deletion","message":"Add spec for barbican secret deletion\n\nChange-Id: I92940d839a38fe848dfa586a9bc2642848078eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/fabcf02c0d376d663792656598d2e32b39fc89bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/fabcf02c0d376d663792656598d2e32b39fc89bc"}]},"branch":"refs/heads/master"},"5e3cb4f599d584d90872d71f8b88e44620352f82":{"kind":"REWORK","_number":4,"created":"2019-07-23 19:36:20.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/95/656895/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/95/656895/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/95/656895/4"}}},"commit":{"parents":[{"commit":"bc4d0c5f89f96bf720db29d0cbd04092a9e06cb7","subject":"Spec Lite: Add location url prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/bc4d0c5f89f96bf720db29d0cbd04092a9e06cb7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-02 22:25:47.000000000","tz":-360},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-07-23 19:35:34.000000000","tz":-240},"subject":"Add spec for barbican secret deletion","message":"Add spec for barbican secret deletion\n\nChange-Id: I92940d839a38fe848dfa586a9bc2642848078eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/5e3cb4f599d584d90872d71f8b88e44620352f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/5e3cb4f599d584d90872d71f8b88e44620352f82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
