)]}'
{"id":"openstack%2Fglance-specs~660263","triplet_id":"openstack%2Fglance-specs~master~I9fa3da97db1064d28e66fc1be42fde1bf85524ab","project":"openstack/glance-specs","branch":"master","topic":"specs/location-uri-prefix","hashtags":[],"change_id":"I9fa3da97db1064d28e66fc1be42fde1bf85524ab","subject":"Spec Lite: Add location url prefix","status":"MERGED","created":"2019-05-21 07:14:10.000000000","updated":"2019-07-08 14:59:55.000000000","submitted":"2019-07-08 14:59:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660263-1562597995123-fd40a304","meta_rev_id":"2edfba89550f9e3cdd916d3b7e76f28ab6aa1511","_number":660263,"virtual_id_number":660263,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-07-08 14:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-08 14:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-08 14:50:42.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 16:50:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-06-03 18:47:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-08 14:50:42.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-07-08 14:59:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4b35eac6d8572742b96eaef9f32375abb48f063","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-21 07:14:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ebae2ab0da674c3141256a760f28f2df8ccf1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 07:25:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/660263/1/check/openstack-tox-docs/c37da39/html/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"23e69180c17c74bfbf8c9da125bcd1a736da6c68","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-06-03 16:50:10.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is a good strategy to make updating the location metadata much less painful.","accounts_in_message":[],"_revision_number":1},{"id":"924edea87accbeff295e222a45e6fea0cd28ca07","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 18:47:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6a331dc5fafb443db347c1cf00bbf16156e024a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-20 04:41:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efdaeacd88d3b88a5add7e22677e6f1b34417d1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 04:56:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/660263/2/check/openstack-tox-docs/0bc36a3/html/ : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"11e5fa4889d9fb9a18a823bfa10bdf9dc7486c4b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 14:49:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0924760d6537190510f5796f963d20edbe1f9ed","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 14:50:42.000000000","message":"Patch Set 2: Workflow+1\n\nBrian and Sean already +2\u0027d","accounts_in_message":[],"_revision_number":2},{"id":"2a8db4f3ca72a41f7c069a3fb8cab373e7580e3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:50:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9a50c0fce0337043cbe630fdb704fba8e3620d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:59:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/660263/2/gate/openstack-tox-docs/b9cb2c9/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"74530cf45ed03dd9926b45e7859233580140b889","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:59:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc4d0c5f89f96bf720db29d0cbd04092a9e06cb7","revisions":{"2b4285cf797d1ed293cedcc78424ce7387cf3138":{"kind":"REWORK","_number":1,"created":"2019-05-21 07:14:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/63/660263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/63/660263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/1"}}},"commit":{"parents":[{"commit":"4e095520a4093e803a24fb7e537435e568947ce8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/4e095520a4093e803a24fb7e537435e568947ce8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-21 07:13:21.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-21 07:14:00.000000000","tz":0},"subject":"Spec Lite: Add location url prefix","message":"Spec Lite: Add location url prefix\n\nChange-Id: I9fa3da97db1064d28e66fc1be42fde1bf85524ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/2b4285cf797d1ed293cedcc78424ce7387cf3138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/2b4285cf797d1ed293cedcc78424ce7387cf3138"}]},"branch":"refs/heads/master"},"bc4d0c5f89f96bf720db29d0cbd04092a9e06cb7":{"kind":"REWORK","_number":2,"created":"2019-06-20 04:41:01.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/63/660263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/63/660263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/63/660263/2"}}},"commit":{"parents":[{"commit":"9656932e7dea300464ede5743f0b82e9b96ab0fc","subject":"Merge \"Spec Lite: Lazy update stores information\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9656932e7dea300464ede5743f0b82e9b96ab0fc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-05-21 07:13:21.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-06-20 04:37:48.000000000","tz":0},"subject":"Spec Lite: Add location url prefix","message":"Spec Lite: Add location url prefix\n\nChange-Id: I9fa3da97db1064d28e66fc1be42fde1bf85524ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/bc4d0c5f89f96bf720db29d0cbd04092a9e06cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/bc4d0c5f89f96bf720db29d0cbd04092a9e06cb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
