)]}'
{"id":"openstack%2Fglance-specs~740980","triplet_id":"openstack%2Fglance-specs~master~I2290356485b5391149cb541322142505136df660","project":"openstack/glance-specs","branch":"master","topic":"g_s-ceph-store-optimization","hashtags":[],"change_id":"I2290356485b5391149cb541322142505136df660","subject":"Optimize Ceph store network usage","status":"ABANDONED","created":"2020-07-14 14:42:59.000000000","updated":"2021-02-03 13:02:11.000000000","total_comment_count":9,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"c662ed311c57aa0ba042e316191d17bc8af370f4","_number":740980,"virtual_id_number":740980,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-20 15:45:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2021-01-21 14:09:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":-1,"date":"2021-01-28 14:55:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 22:37:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-23 19:15:47.000000000","updated_by":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"reviewer":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"state":"REVIEWER"},{"updated":"2020-08-04 19:44:49.000000000","updated_by":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"reviewer":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"state":"REVIEWER"},{"updated":"2020-08-18 14:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 14:09:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-28 14:55:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"94824259f5417f3da11a7879d27c7c90c542eff3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-14 14:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"521f390035a5245a01f8d10481b2900f23a9d54d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-14 14:52:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cad3507957f0883a0daccf4d7a7aa705a6b1165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 15:03:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd830a22e0e7400487b5b69b3861020e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"261a99fa84b34fe40303d509272819219e371a9f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-18 22:37:31.000000000","message":"Patch Set 2: Code-Review+2\n\nAll good things to work towards improving.","accounts_in_message":[],"_revision_number":2},{"id":"49f06cbe5a5e5340d35c066c669ad697b8e99063","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-07-23 19:15:47.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nLGTM but I still have trouble seeing exactly what the \"RADOS Optimization\" mean, can you add some example, or some references to radoslib/ceph documentation about what it mean ?","accounts_in_message":[],"_revision_number":2},{"id":"8b571f74efca68588e0ae8eace74cd5805bf8d8a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-30 14:54:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9cf548d9d2b3aa720889f3e3dc59f89ac7c8e91","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-18 14:39:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0eb668187729358fed9a4637dce2a89fa8c7df87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:47:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451c141a27fb40a59b19891147f4383f : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0ad28c4e53ccef7e47690cada65234dc2f6ff93e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-01-19 12:05:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bfd8acf59a83e097efe8c063f0393f61c8f55831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 12:11:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6b0b8fd09f46d2bdb152a6910607ff : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0035b07e3964a550fd5729460b7f1ab55cd2673f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-01-19 12:50:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4b8407d50227b66973447b9b4afaae149a025b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 13:10:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5790c39122e24ac6b3dc896b976931c6 : FAILURE in 5m 25s","accounts_in_message":[],"_revision_number":5},{"id":"57a93d26e5b3ac6ce95b3885ee2edd7a655ab3eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-01-20 15:35:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"75e284d78d15ef6bd7820a1ec497609fc8a8f49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 15:45:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a44e5c8ad44fc89446835a0935579a : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":6},{"id":"bf06716464a46fbfd71103653894a23303753761","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-21 14:09:49.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nOne question and some typos noted inline.\n\nCinder recently announced a policy restricting what versions of Ceph are \"supported\":\nhttps://opendev.org/openstack/cinder/commit/2f5179147c36308c6d1e6a29d5f98b2124e6346c\n(mainly because we\u0027re only doing CI testing with whatever version is supplied by the ceph devstack plugin, which tends to be fairly recent).  I wonder whether we should do the same for glance?","accounts_in_message":[],"_revision_number":6},{"id":"7439615eced312d0eaff125a1fe934cf774ae6bd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-28 14:55:02.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c662ed311c57aa0ba042e316191d17bc8af370f4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-03 13:02:11.000000000","message":"Abandoned\n\nLooks like thim missed the boat too","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d95147355d184c045dd08a03434286b021c4dc82","revisions":{"d5c83fc5731eada8475aa1b0eedbcfcddfac0645":{"kind":"REWORK","_number":1,"created":"2020-07-14 14:42:59.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/740980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/80/740980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/1"}}},"commit":{"parents":[{"commit":"576a8e20cdc06e0a0ee9ba45f2b5bc977b0f7dea","subject":"Add copy-unowned-image spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/576a8e20cdc06e0a0ee9ba45f2b5bc977b0f7dea"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"subject":"Optimize Ceph store network usage","message":"Optimize Ceph store network usage\n\nChange-Id: I2290356485b5391149cb541322142505136df660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d5c83fc5731eada8475aa1b0eedbcfcddfac0645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d5c83fc5731eada8475aa1b0eedbcfcddfac0645"}]},"branch":"refs/heads/master"},"4f6fd2a87243336ea2326142947a7c3e316e19d4":{"kind":"REWORK","_number":2,"created":"2020-07-14 14:52:01.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/740980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/80/740980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/2"}}},"commit":{"parents":[{"commit":"576a8e20cdc06e0a0ee9ba45f2b5bc977b0f7dea","subject":"Add copy-unowned-image spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/576a8e20cdc06e0a0ee9ba45f2b5bc977b0f7dea"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:51:49.000000000","tz":60},"subject":"Optimize Ceph store network usage","message":"Optimize Ceph store network usage\n\nChange-Id: I2290356485b5391149cb541322142505136df660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/4f6fd2a87243336ea2326142947a7c3e316e19d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/4f6fd2a87243336ea2326142947a7c3e316e19d4"}]},"branch":"refs/heads/master"},"c697c6db686d9b1f4e53ebfa6f2b0047c7798033":{"kind":"REWORK","_number":3,"created":"2020-08-18 14:39:12.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/740980/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/80/740980/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/3"}}},"commit":{"parents":[{"commit":"576a8e20cdc06e0a0ee9ba45f2b5bc977b0f7dea","subject":"Add copy-unowned-image spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/576a8e20cdc06e0a0ee9ba45f2b5bc977b0f7dea"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-18 14:38:59.000000000","tz":0},"subject":"Optimize Ceph store network usage","message":"Optimize Ceph store network usage\n\nChange-Id: I2290356485b5391149cb541322142505136df660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/c697c6db686d9b1f4e53ebfa6f2b0047c7798033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/c697c6db686d9b1f4e53ebfa6f2b0047c7798033"}]},"branch":"refs/heads/master"},"62b391377acfc3a113a9ce7c14dcbc6c5b2ffd19":{"kind":"REWORK","_number":4,"created":"2021-01-19 12:05:01.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/740980/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/80/740980/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/4"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-01-19 12:04:35.000000000","tz":0},"subject":"Optimize Ceph store network usage","message":"Optimize Ceph store network usage\n\nChange-Id: I2290356485b5391149cb541322142505136df660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/62b391377acfc3a113a9ce7c14dcbc6c5b2ffd19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/62b391377acfc3a113a9ce7c14dcbc6c5b2ffd19"}]},"branch":"refs/heads/master"},"9e4c309cb03491fb70d7ba1488d3440cae8a8fed":{"kind":"REWORK","_number":5,"created":"2021-01-19 12:50:48.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/740980/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/80/740980/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/5"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-01-19 12:50:33.000000000","tz":0},"subject":"Optimize Ceph store network usage","message":"Optimize Ceph store network usage\n\nChange-Id: I2290356485b5391149cb541322142505136df660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9e4c309cb03491fb70d7ba1488d3440cae8a8fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9e4c309cb03491fb70d7ba1488d3440cae8a8fed"}]},"branch":"refs/heads/master"},"d95147355d184c045dd08a03434286b021c4dc82":{"kind":"REWORK","_number":6,"created":"2021-01-20 15:35:04.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/740980/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/80/740980/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/80/740980/6"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-14 14:42:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-01-20 15:34:51.000000000","tz":0},"subject":"Optimize Ceph store network usage","message":"Optimize Ceph store network usage\n\nChange-Id: I2290356485b5391149cb541322142505136df660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d95147355d184c045dd08a03434286b021c4dc82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d95147355d184c045dd08a03434286b021c4dc82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
