)]}'
{"id":"openstack%2Fglance-specs~763740","triplet_id":"openstack%2Fglance-specs~master~I2139af70efb497587f99a435f852e9dcd07cedb5","project":"openstack/glance-specs","branch":"master","topic":"expose-task-show-api","hashtags":[],"change_id":"I2139af70efb497587f99a435f852e9dcd07cedb5","subject":"New Image API /v2/images/{id}/tasks for task information","status":"MERGED","created":"2020-11-23 06:32:33.000000000","updated":"2021-02-12 15:33:12.000000000","submitted":"2021-02-12 15:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":108,"unresolved_comment_count":29,"has_review_started":true,"submission_id":"763740-expose-task-show-api","meta_rev_id":"c00b71f431d82b2b36bcd59e07255b12a19b8693","_number":763740,"virtual_id_number":763740,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 15:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-02-11 14:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-02-10 15:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-02-12 15:25:30.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 06:49:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 13:48:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-07 18:26:04.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-01-19 13:00:52.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2021-02-04 14:44:02.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"0b9db6a1025c7644ee928dfff84fc2e85395b5d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-11-23 06:32:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"841aab16e5674fa2c733720501644bd6c2655973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 06:49:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b369cd9d864290b1b65bf038456907 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b9d595a31641b0b1c6e955db5ddf5cf021673f37","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-10 13:48:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI agree that we need to get this info to users, but at first glance, this looks like an abuse of the discovery endpoint.  It would be helpful to have some more discussion of why you reject the alternative.","accounts_in_message":[],"_revision_number":1},{"id":"34c793c1c6c8a025b9b2a165c0be4f4bfbf5ab1d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-10 14:16:47.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThank you Brian, please have a look at inline answere","accounts_in_message":[],"_revision_number":1},{"id":"d5e6ec436c0dad80a1b2779b66516ccded71dc1a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-10 14:36:07.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nReplies inline.","accounts_in_message":[],"_revision_number":1},{"id":"ef613bec07926d26e30ccfd3fcf87bbc531313c6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-10 14:42:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1213ef943b990a114a5f0e32ccf36545b8ffc462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-14 07:24:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ba0ccfc120b7adac687f5c13fe383f2032a47b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 07:34:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e22329a71ed4aab84a94bf54c65bc9f : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"4da2a1138669266e2816eb03ce63223f5b546a5a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-12-17 16:11:14.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSuggestion inline for a slightly different way to handle this, though I\u0027m not necessarily against the current proposal.","accounts_in_message":[],"_revision_number":2},{"id":"2443ef4ec00490173fec26309ab99ddb839eaf37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-18 07:11:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"664396237bc0cb3ad001210450d743e56151cdca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-18 07:17:31.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThank you Brian, updated specs as per your inputs","accounts_in_message":[],"_revision_number":3},{"id":"1be1e3472cfe1055220f8dbafd4ab39093d6dded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 07:21:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856f2f9f28fc46358cef638b3fda6bc6 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"e075c61685fca9ab263836a269b0c16a08d3e4e0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-07 16:52:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)\n\nA few wording nits noted inline, plus a bigger suggestion about what the response should look like (though I\u0027m not sure how good a suggestion it is).  But I like the overall direction to make this a sub-resource of an image in the API, it\u0027s much more user-friendly.","accounts_in_message":[],"_revision_number":3},{"id":"3c851e7eefd1859b8a25120c942c3938df720599","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-07 17:01:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1da30a6db14dccc627b6dd98fc3879469020004","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-07 18:26:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nI\u0027m not entirely sure why this is necessarily better than making the actual tasks API do the right thing, but that\u0027s fine. I\u0027m definitely in favor of being able to access this information from the client regardless of how :)","accounts_in_message":[],"_revision_number":3},{"id":"a35f7eebd862231d6ac58b9e98c4db1771142385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-08 04:33:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa938d61300dd9967be397496024ccb527d4518c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-08 04:34:56.000000000","message":"Patch Set 4:\n\n(14 comments)\n\nThank you Brian and Dan for reviews, made changes as per your suggestions.","accounts_in_message":[],"_revision_number":4},{"id":"5405f5e142c583075ca45e5c6163fe6f7ed3fc69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:40:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff252872b4f4307b2c8c8540a4a5bc1 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d1029b8130ca7349d7438067232a2433b6e07ff3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-08 23:14:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nA few questions inline.","accounts_in_message":[],"_revision_number":4},{"id":"2e2f0eea6b69dd0f3dd966db8a75da481097decf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-08 23:23:59.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0ef55b3809a1ec7a54d387fe53daa0518f08a602","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-09 23:16:13.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nA few more thoughts inline.","accounts_in_message":[],"_revision_number":4},{"id":"f878e83a7e993ed0032669a34a19f2a468d8b08a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-11 05:12:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fed4a74b9bd9a6f96356b0466cb196b7df852759","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-11 14:44:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"51803450e0a9833f750b658a3b028cb714283141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-12 06:21:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8cffd12adf9b946e6461292d905e28ba563f6077","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-12 06:23:07.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nThank you for review dansmith and rosmaita, updated the specs as per your suggestions.","accounts_in_message":[],"_revision_number":5},{"id":"ffb4fe6e0958149207f667d0b9822d35c8fd8713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 06:28:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b7c5e41c614e81ad3b16a07b388f24 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":5},{"id":"60046595ffbf67c534875d5a476aaeacfc64f32a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-12 21:14:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(11 comments)\n\nCouple more things. I will revise this for Abhi.","accounts_in_message":[],"_revision_number":5},{"id":"faa757a539930f6625ed100a976b74676077a0bc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-12 22:04:27.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nA few nits inline.  Otherwise, the only thing missing is a JSON schema for the response.  But that can probably be dealt with later when you decide whether to change the task schema to add the new fields, or use a new one for this response.  (The tasks list schema shows a summary for each task, not the task detail that you have here.)","accounts_in_message":[],"_revision_number":5},{"id":"25c2d16f932b4e8133069ef2df318c4b2bfb8276","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-12 23:08:29.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nI\u0027ll leave the schema for a later update. Seems like a fine thing to amend the spec with later.","accounts_in_message":[],"_revision_number":5},{"id":"9afc1dfe3fd5417a22e8c25465b2c0855cb06e61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-12 23:09:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d0be6a64b5756c828a7b6fc7d139c07c8dbe1b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 23:20:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e365793ed7f347d2997335122c1a76f1 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":6},{"id":"0e85f71ef6294ef4a7b69ac72548897c3acf21e6","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-01-19 13:00:52.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5ddf979b923b3c47ef15f419814a5234b996ac17","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-19 18:59:41.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nWe need a plan for deleting expired tasks, possibly by moving (or copying) the lazy deletion currently in the list tasks code to the code that handles this call.","accounts_in_message":[],"_revision_number":6},{"id":"93c26837daf04f068e4609d3d1573c51628a4946","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-19 20:00:54.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5c78009474c82f6eb462e37206441b6525a8ed83","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 04:59:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1151ba98d8a962600f81834329eaade82e12515a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 04:59:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5b92fddd6d260e49af181558547db5f61b3d6a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 05:08:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20131d9b94aa4a5fb1c24d05bb9eb5f2 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":7},{"id":"64d3561661eda9f57cb10c712947e9e724642fc5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-28 16:06:22.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\nContent is fine.  Looks like PS7 introduced a bunch of typos that weren\u0027t in PS6, but the intent is clear enough.","accounts_in_message":[],"_revision_number":7},{"id":"9b10d86d16c2dde9b40bee1c87bb7892d8deb64d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-28 17:44:05.000000000","message":"Patch Set 7: Code-Review+2\n\n(6 comments)\n\nAbhi, I think you clobbered some of my fixes from PS6. However, I think this is close enough, let\u0027s just land it so we are clear we\u0027re going to get this done this cycle and one of us can follow up with fixes to the words here.","accounts_in_message":[],"_revision_number":7},{"id":"ac6d88f0c1324b3f5147bea894634178aa124b12","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-04 13:41:05.000000000","message":"Patch Set 7:\n\n(6 comments)\n\nThank you Dan, I have made changes as per your suggestions.","accounts_in_message":[],"_revision_number":7},{"id":"9d2e7fdb1fc71ff2d57551ec8ae2b8fa517c89bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-04 13:41:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0abecb4edff77c67ff90c22ca656dbe4023647c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 13:48:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef52c23b66ff450ead862280fb68238e : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":8},{"id":"013f38bee5edb10c48a1b72740e5f11deba5dd0e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-04 14:20:41.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks for making those typo edits, let\u0027s merge this!","accounts_in_message":[],"_revision_number":8},{"id":"d7d8efe6150a6756e6c494ed8e24f4e4b639e719","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-04 14:44:02.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"54d31bae5448db811f565ff756e282fc8fb8e308","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-04 14:52:45.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"144609b128a2840c8bf308192b38979c283ea74d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-04 16:51:22.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f5218218c2aec0887e8c164622e0d553590b27bb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-04 17:22:00.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e264c8462eabe2777378fb919d115b932b8ff1cc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-04 18:10:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f5e545b2c74668edae969677f03c8e9f9fd92f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-09 05:34:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ddbfed71ac028321452b0bdeffb7ee12b449f3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 05:41:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb057607278e47f7b4c30248c3316844 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":9},{"id":"91334ca290e5a46d316f1462d2ec07a7b297e70f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-10 15:35:29.000000000","message":"Patch Set 9: Code-Review+2\n\nStill good with this and hope we can still get it in this cycle.","accounts_in_message":[],"_revision_number":9},{"id":"2b6d23882fe9cef311320dc797e1932812b46da3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-11 14:19:43.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nOne comment inline about something that could be clarified on a followup, but I think the intention is clear enough and I like the overall proposal.","accounts_in_message":[],"_revision_number":9},{"id":"1a49c14fda217af72ca790a412a5856295b03cb1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-12 04:49:33.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nThank you Brian, IF you don\u0027t have any objection then I will approve this spec EOD today.","accounts_in_message":[],"_revision_number":9},{"id":"4798b4982d0e660001394df0287d2323185cb6f8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-12 15:25:30.000000000","message":"Patch Set 9: Workflow+1\n\nI will address Brian\u0027s comment in follow up.\nThank you all for review!","accounts_in_message":[],"_revision_number":9},{"id":"3efaf42bfcb3bde1bd2994acc4a51e0e858679b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:25:44.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"07435482d5eb64d86ed19cf6807b4d59e69db9c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:31:09.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0f34a469e74f08941debea0dd1fdb9 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":9},{"id":"a9042c394998456d90071911b420ef92872f30a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:31:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"c00b71f431d82b2b36bcd59e07255b12a19b8693","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:33:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/2d7151023a61439e9c0f76c001f1aa1b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e518123237e7882d4b31b5ba8bd068cbb5e28a89","revisions":{"528ce40d9dd5f794c8fd39e54e6d42812ebc9914":{"kind":"REWORK","_number":1,"created":"2020-11-23 06:32:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/1"}}},"commit":{"parents":[{"commit":"d7bbf477959c0bd02f94f4896bc3f224cd313e34","subject":"Create specs directory for Wallaby cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d7bbf477959c0bd02f94f4896bc3f224cd313e34"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:53.000000000","tz":0},"subject":"[Spec lite] Discovery API for task","message":"[Spec lite] Discovery API for task\n\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/528ce40d9dd5f794c8fd39e54e6d42812ebc9914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/528ce40d9dd5f794c8fd39e54e6d42812ebc9914"}]},"branch":"refs/heads/master"},"7b107a9a5937cd70627495345a999301e77d2754":{"kind":"REWORK","_number":2,"created":"2020-12-14 07:24:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/2"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-12-14 07:22:54.000000000","tz":0},"subject":"[Spec lite] New Endpoint /v2/messages/{id} for task show API","message":"[Spec lite] New Endpoint /v2/messages/{id} for task show API\n\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/7b107a9a5937cd70627495345a999301e77d2754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/7b107a9a5937cd70627495345a999301e77d2754"}]},"branch":"refs/heads/master"},"d0e856fd062a77333d368fb0ebd4c8c57370e129":{"kind":"REWORK","_number":3,"created":"2020-12-18 07:11:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/3"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-12-18 07:09:42.000000000","tz":0},"subject":"New Image API /v2/images/{id}/messages for task information","message":"New Image API /v2/images/{id}/messages for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d0e856fd062a77333d368fb0ebd4c8c57370e129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d0e856fd062a77333d368fb0ebd4c8c57370e129"}]},"branch":"refs/heads/master"},"e4e52b4d21730c6701f0f92fd67cfc5a864fdfab":{"kind":"REWORK","_number":4,"created":"2021-01-08 04:33:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/4"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-01-08 04:29:13.000000000","tz":0},"subject":"New Image API /v2/images/{id}/messages for task information","message":"New Image API /v2/images/{id}/messages for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e4e52b4d21730c6701f0f92fd67cfc5a864fdfab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e4e52b4d21730c6701f0f92fd67cfc5a864fdfab"}]},"branch":"refs/heads/master"},"6ab00d35d79d324d252eff0579ba745070d14b73":{"kind":"REWORK","_number":5,"created":"2021-01-12 06:21:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/5"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-01-12 06:16:33.000000000","tz":0},"subject":"New Image API /v2/images/{id}/tasks for task information","message":"New Image API /v2/images/{id}/tasks for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6ab00d35d79d324d252eff0579ba745070d14b73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6ab00d35d79d324d252eff0579ba745070d14b73"}]},"branch":"refs/heads/master"},"63fed859ca35aa068ccdc6effdcbae9c260d75e6":{"kind":"REWORK","_number":6,"created":"2021-01-12 23:09:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/40/763740/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/6"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 00:06:45.000000000","tz":-480},"subject":"New Image API /v2/images/{id}/tasks for task information","message":"New Image API /v2/images/{id}/tasks for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/63fed859ca35aa068ccdc6effdcbae9c260d75e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/63fed859ca35aa068ccdc6effdcbae9c260d75e6"}]},"branch":"refs/heads/master"},"cb4fab6e3511a4bbd40b098f56c939a0c842d45a":{"kind":"REWORK","_number":7,"created":"2021-01-22 04:59:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/7"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-01-22 04:53:37.000000000","tz":0},"subject":"New Image API /v2/images/{id}/tasks for task information","message":"New Image API /v2/images/{id}/tasks for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/cb4fab6e3511a4bbd40b098f56c939a0c842d45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/cb4fab6e3511a4bbd40b098f56c939a0c842d45a"}]},"branch":"refs/heads/master"},"124ff2ce3dc9bc123202521829b75fa5db879f7e":{"kind":"REWORK","_number":8,"created":"2021-02-04 13:41:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/8"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-04 13:33:51.000000000","tz":0},"subject":"New Image API /v2/images/{id}/tasks for task information","message":"New Image API /v2/images/{id}/tasks for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/124ff2ce3dc9bc123202521829b75fa5db879f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/124ff2ce3dc9bc123202521829b75fa5db879f7e"}]},"branch":"refs/heads/master"},"e518123237e7882d4b31b5ba8bd068cbb5e28a89":{"kind":"REWORK","_number":9,"created":"2021-02-09 05:34:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/40/763740/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/40/763740/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/40/763740/9"}}},"commit":{"parents":[{"commit":"6de8257026039e7cc71afea65d459a00beea0f85","subject":"Fix redirects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6de8257026039e7cc71afea65d459a00beea0f85"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-11-23 06:31:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-09 05:26:08.000000000","tz":0},"subject":"New Image API /v2/images/{id}/tasks for task information","message":"New Image API /v2/images/{id}/tasks for task information\n\nblueprint: messages-api\nChange-Id: I2139af70efb497587f99a435f852e9dcd07cedb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e518123237e7882d4b31b5ba8bd068cbb5e28a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e518123237e7882d4b31b5ba8bd068cbb5e28a89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
