)]}'
{"id":"openstack%2Fglance-specs~947423","triplet_id":"openstack%2Fglance-specs~master~I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3","project":"openstack/glance-specs","branch":"master","topic":"bp/set-size-on-upload","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-05-29 12:54:10.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-04-24 15:38:54.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2025-06-09 14:41:29.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-06-09 14:41:29.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-06-09 14:41:29.000000000","reason":"Change was submitted"},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2025-06-09 14:41:29.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-06-09 14:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-06-09 14:41:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3","subject":"Add spec to provide image size in advance","status":"MERGED","created":"2025-04-16 09:06:59.000000000","updated":"2025-06-09 14:42:23.000000000","submitted":"2025-06-09 14:41:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":86,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"947423-bp/set-size-on-upload","meta_rev_id":"1ad5512abd15d0ff2c0a8a193ea71cba55cd73c5","_number":947423,"virtual_id_number":947423,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 14:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-05-29 12:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-06-09 14:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2025-06-09 14:35:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 09:07:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2025-04-16 09:07:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-04-16 09:07:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2025-04-16 09:07:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-04-16 09:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-21 13:40:40.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-04-21 14:44:11.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-05-19 14:29:10.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-05-22 19:40:45.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"d81c167658143d4d15cb2ea36333b7d72f1070e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-16 09:06:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49b5b3b1410dad48ba9c7184b69775795960ba37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 09:12:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba65b1f3cbd470ba5ff814f493ef8dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33579a5e8e8e47bcac0717b339b08422 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b30b68c4a7f156e158024e14d09ab8dcd6d5e480","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-16 18:38:10.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ed0e2efc272976b38471a83a966be26cd5f16435","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-17 06:09:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ada3ef8b1dc44e3a1878ea24db8e93231e319d60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-17 06:13:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b33dc4ec935e09f332733b92dd7ea7678c34f957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 06:17:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/281095050e704c6590255475d8768c03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e675d256a640bea4c7fbbe45bf4fc5 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"7f94f62b008e8d991e52fc29b3f71eea0ea943f9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-04-17 08:00:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b95b66cd3fdae7ff972658216ae96a4f6af273b5","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-17 23:02:17.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"929f3e9f553fb010a688a67c7b2930405185eef6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-18 01:59:08.000000000","message":"Topic bp/by-pass-chunked-transfer removed","accounts_in_message":[],"_revision_number":2},{"id":"0b056b15c6253e987dd272cba1288ae277a28099","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-18 02:00:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd83cd987ceab3c27e0a58f52cce567527f5516e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 04:54:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"abb99fde48621fb6abd693957c4c9b10cc97fc93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 04:56:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499a602bcb164811aa9cc3981438692c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f363c0b6b6b143b9912e2447f442e1e7 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6606d66fd339697879dac928afdd9d07114afc35","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-21 13:40:40.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eda745a933c9840e28e39f93675056422e678348","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 14:06:56.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0e1ffa098ab5b8f614e55c5fc944125ebaba5030","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-21 14:12:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb1eef5c8269cecddf97e5c80107c778d4d433aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 14:15:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"91a3b9baa679185ff1fcdbd6a3862af13d4a75b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 14:24:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996f83420f1545dfa2b87c2bfe3fd837\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0265256afd294db19e134a0b0a7ac8f1 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":4},{"id":"9f2c91c3744095dde5417ea16fbf33c67071fd76","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 14:24:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2af00ac1ef1823ad6fccd5e14b034aed9445cb78","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-21 14:39:26.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"afc06a85380476737837cb6b2d1a9f1e5f2ef2a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-21 14:44:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd388101537bee17219764a1c85800143b650d69","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-21 15:26:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3dea8268e6545c33fdb92c4d001a899e456e4479","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-21 15:42:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"addfa6a3ef72204f7f5b815920aa8bcca74a5f74","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-23 02:25:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e9389900a5e17ebf050398d1e109b29652af875","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-23 13:33:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e7b54ca1b0a5c3629e202dbcd7d2f8a89893a986","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-23 20:23:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d5046408d035b2ce7db0f895892e62a80820622","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-23 21:20:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5023f4adf6b0107d98f26389f62de6e4726c026e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-23 22:05:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8885e76f9918654428cff208906626d0e66770cd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-23 22:27:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c9848840cae5c3a3ded074bf151463e90b62538","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-23 22:57:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6dfdd9f695962e79489df7b3aa42d1e2632259f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-24 13:49:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9274000c8a9fc5616ce285eecf57d1c7ad00ec15","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-24 14:52:16.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"9ad679154f77790d8198e651fef2d064e416dd1d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-24 15:18:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d1fca8abf2c7b2eae1ea5c48346348dded4440fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-24 15:34:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56595488cfef4d194d0590b60997e67bea009007","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-24 15:38:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"69d8451109cc0b32efa8e0fbfeb521f3b7d0f174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-25 08:11:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0cf3157605f31934d812a065fc071e71526bd93e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 08:20:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/035314a5ee8043acbbfefa9be8ba8717\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a190ae786afd44fe85963a6ac8bbbd67 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"49288630327e156b6697586adb9e1682f39b6973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-25 14:32:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c08335720768fcdf0a553d9d97da17157624bf62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 14:37:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/059d38eee44349cca4068bfc65f80486\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76dfe9236c8f4ea6949767893e843bb4 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":6},{"id":"f42d4731f0068c32fd2818125ab7c8e3c653dd9e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-25 15:01:14.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"60217de218a6e12981321e8f42d1a18f4fcae3a5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-25 15:16:48.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5930b0344b032787269847e5e7a2234de00cb236","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-25 15:25:14.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e7165882d03448b9b3ce448f791440676125d625","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-25 15:48:04.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"829b611ebbbf6ba336f1a5a2eb70c2e5e7fb94c5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-04-25 16:20:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3e707980feeaebaa367f2b3e034ceb429f08cc9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-28 06:38:21.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c6be4b6d7d9d2ce70092c5699b1a9024e24f756b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-28 06:38:29.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e515c905080fe11ac0aa911cb69b96352ea4fcb7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-04-28 06:38:59.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b232b300a01fb7d5172adc3362ee11e3977a06de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 06:45:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bee3238eec94c298a7fcef466132d5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83628c8dee7946eaa1383c7d0734d4f7 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c24b29cbd2c93e3d94edf74ca3ffdbf7dba0af37","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-05 16:14:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6bd0d895c1cd44126f9d2a50054884fa971e5e9f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-07 08:13:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c2a2df9245e755f65ed8a430303a337ecd48069b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-07 13:34:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"afbf1e5032f6714a8637f6e475bc1d6f9fbc0bd2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-07 14:48:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"08d94fd9e7f845fa50a7cba5372aaaaf596ed942","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-07 14:54:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9378253c2d19ada648ceba35e1c7e75376913e03","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-07 15:00:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e7a9cd6edb3349deb4be3066a32acaf13239c52d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-07 15:13:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"083167d99c4016fdaeb2a88df1814eb9df03642d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-07 15:26:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"64ebd2789b67a22cdf62e2fd24d7ac6cb0a1390a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-07 15:39:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8508e735fc4244aeed2801c6f4d5b423d7f0ae0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-08 05:59:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c95b0c186fa6816a009b7c11a2cf4577aa73a2c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-14 14:44:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"ffd9c02d58e059fe8b506e854bc2d0aa7bb2d79c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 14:54:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b87c64eff5c434b8267c7221a737bbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5a7f8bda1f43c69baeceb2e64e49ce : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":8},{"id":"df217a479fc77e03ba87793dca808f723d6fd17c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-22 16:03:15.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cc0484bb83698481583593a7301cf2650905e18b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-22 17:19:06.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"69b5d01ceab4d74f9ab074d200481bc85e7c8fcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-22 17:19:58.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"20ce1ba178984a4b1a682e1065d7d7f752b19340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:23:02.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63dfa5a9c3554a5a870d2d2dae71b3e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e85c9d7372471192c993c07e5789d5 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":9},{"id":"6dd2d134dfe7c28687c4a4bdf78262aab5b85666","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-22 19:40:45.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"97a156986e56fbf52b3e09063b9db169d63bf1a7","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-05-26 07:39:18.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"267d4bc8e5e5f63038f50c3776255e4c8bb5686d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-28 15:33:42.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9a6a59a3cf547ba3cf8e137bd5fccda480a10041","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-28 15:37:25.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0d46e2546e345ab84eb63965289fe80314db8135","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-28 15:52:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fb8d62cbe2c1892e225143f2e2a35bb7561289b8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-28 16:08:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cba7e061fa8bb1a149cb92b459770a70e428e5aa","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-05-28 16:16:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"22d1946da2ec48f9538a499ddfce76d6519a9120","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-29 05:04:47.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"833cb0407c1b2cb81177d9fb966652e7a955a2c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-29 05:05:59.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"af9e66c1d1ca5c08e0cb833e25740b171935b484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 05:55:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3c14a3ab09444fc93ac3267239335a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e79de49a076484899191d3afe02225d : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":10},{"id":"8d7c973982d4162b5c49b3c1ccc83d9e1e2f5ebc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-29 12:54:10.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"20aca7ef78ab6c93efd1b7cc8edf7b4ea9ba5173","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-29 12:57:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e9965fc9c0c4df05c6a4a153ad63fd161f5eac89","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-09 14:35:39.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a212cf2c8c86327384363f7ecb2180e13b383331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 14:35:48.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"abb3ca4aca0d12567fd95d0860aaf623b1281a46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 14:41:28.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e2822818df42d59c4b670af971f55a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66285494280844e2ba46d3e9aa83ed74 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":10},{"id":"254cd36b8fa1ee4eef73b037685e65572227aaa9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 14:41:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1ad5512abd15d0ff2c0a8a193ea71cba55cd73c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 14:42:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464ec15dc33f47f5ae45f78124c7b510\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/366bf353f98849f58d86645b5d424f84 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e0eb7ac13429931622884f0add9548571988c1bf","revisions":{"0c3cf8a4516e8d1ab5a22ef56e63f7a2caea55b9":{"kind":"REWORK","_number":1,"created":"2025-04-16 09:06:59.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/1"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"subject":"Add spec to bypass chunked transfer of data","message":"Add spec to bypass chunked transfer of data\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/0c3cf8a4516e8d1ab5a22ef56e63f7a2caea55b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/0c3cf8a4516e8d1ab5a22ef56e63f7a2caea55b9"}]},"branch":"refs/heads/master"},"ead01b64d1973d87470dac570362d9ee12203dd4":{"kind":"REWORK","_number":2,"created":"2025-04-17 06:13:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/2"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-17 05:54:29.000000000","tz":0},"subject":"Add spec to bypass chunked transfer of data","message":"Add spec to bypass chunked transfer of data\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/ead01b64d1973d87470dac570362d9ee12203dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/ead01b64d1973d87470dac570362d9ee12203dd4"}]},"branch":"refs/heads/master"},"58feea65a4527162b27ff8d4438f55b657f408c9":{"kind":"REWORK","_number":3,"created":"2025-04-21 04:54:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/3"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-21 04:34:34.000000000","tz":0},"subject":"Add spec to bypass chunked transfer of data","message":"Add spec to bypass chunked transfer of data\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/58feea65a4527162b27ff8d4438f55b657f408c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/58feea65a4527162b27ff8d4438f55b657f408c9"}]},"branch":"refs/heads/master"},"697e1760b3383c8b9e0ebebe104b5d291e1a836c":{"kind":"REWORK","_number":4,"created":"2025-04-21 14:15:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/4"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-21 13:55:54.000000000","tz":0},"subject":"Add spec to bypass chunked transfer of data","message":"Add spec to bypass chunked transfer of data\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/697e1760b3383c8b9e0ebebe104b5d291e1a836c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/697e1760b3383c8b9e0ebebe104b5d291e1a836c"}]},"branch":"refs/heads/master"},"087a0fad196330dac8363885e9e803782cd4e9f5":{"kind":"REWORK","_number":5,"created":"2025-04-25 08:11:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/5"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-25 07:51:41.000000000","tz":0},"subject":"Add spec to provide image size in advance","message":"Add spec to provide image size in advance\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/087a0fad196330dac8363885e9e803782cd4e9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/087a0fad196330dac8363885e9e803782cd4e9f5"}]},"branch":"refs/heads/master"},"1e16b9242c0da02ba77d7ba155362c2fc53b1db5":{"kind":"REWORK","_number":6,"created":"2025-04-25 14:32:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/6"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-25 14:12:50.000000000","tz":0},"subject":"Add spec to provide image size in advance","message":"Add spec to provide image size in advance\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/1e16b9242c0da02ba77d7ba155362c2fc53b1db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/1e16b9242c0da02ba77d7ba155362c2fc53b1db5"}]},"branch":"refs/heads/master"},"b507d16f5ab274aad903244e7739f5edccdd414b":{"kind":"REWORK","_number":7,"created":"2025-04-28 06:38:21.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/7"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-28 06:18:39.000000000","tz":0},"subject":"Add spec to provide image size in advance","message":"Add spec to provide image size in advance\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/b507d16f5ab274aad903244e7739f5edccdd414b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/b507d16f5ab274aad903244e7739f5edccdd414b"}]},"branch":"refs/heads/master"},"50b79b6a02adda04dc053bd2614c50bc8fc48730":{"kind":"REWORK","_number":8,"created":"2025-05-14 14:44:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/8"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-14 14:23:54.000000000","tz":0},"subject":"Add spec to provide image size in advance","message":"Add spec to provide image size in advance\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/50b79b6a02adda04dc053bd2614c50bc8fc48730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/50b79b6a02adda04dc053bd2614c50bc8fc48730"}]},"branch":"refs/heads/master"},"40b2ca52d1e7f056d3932f69d8defaf65b7f6197":{"kind":"REWORK","_number":9,"created":"2025-05-22 17:19:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/9"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-22 16:58:32.000000000","tz":0},"subject":"Add spec to provide image size in advance","message":"Add spec to provide image size in advance\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/40b2ca52d1e7f056d3932f69d8defaf65b7f6197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/40b2ca52d1e7f056d3932f69d8defaf65b7f6197"}]},"branch":"refs/heads/master"},"e0eb7ac13429931622884f0add9548571988c1bf":{"kind":"REWORK","_number":10,"created":"2025-05-29 05:05:59.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/947423/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/23/947423/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/23/947423/10"}}},"commit":{"parents":[{"commit":"a3a279d4247fd91a50df2a27d160aab2f7147d44","subject":"Add 2025.2 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a3a279d4247fd91a50df2a27d160aab2f7147d44"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-04-16 08:47:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-05-29 04:43:53.000000000","tz":0},"subject":"Add spec to provide image size in advance","message":"Add spec to provide image size in advance\n\nChange-Id: I9c53d93c518fd7f1032e34bd8a0b17a0ee4c3ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e0eb7ac13429931622884f0add9548571988c1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e0eb7ac13429931622884f0add9548571988c1bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
