)]}'
{"id":"openstack%2Fglance-specs~962245","triplet_id":"openstack%2Fglance-specs~master~I4fe29438a1cae8e705caa1337026c970b564ac72","project":"openstack/glance-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"35491":{"account":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"last_update":"2026-06-24 06:30:05.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-10-29 05:23:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-24 06:30:05.000000000","reason":"Change was abandoned"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-06-24 06:30:05.000000000","reason":"Change was abandoned"},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2026-06-24 06:30:05.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-24 06:30:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4fe29438a1cae8e705caa1337026c970b564ac72","subject":"Add new container-registry import method","status":"ABANDONED","created":"2025-09-25 06:15:28.000000000","updated":"2026-06-24 06:30:05.000000000","total_comment_count":35,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"47f96c3259e0f75aa6b0d950d5bcbb8bc2ba7692","_number":962245,"virtual_id_number":962245,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-10 18:57:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2026-01-05 13:21:22.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 06:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 12:41:12.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-09-30 15:45:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2025-10-10 15:41:37.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-10-15 09:20:26.000000000","updated_by":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"reviewer":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"state":"CC"},{"updated":"2025-10-15 09:32:23.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"CC"}],"messages":[{"id":"18d678bb65d60e0a637acf8f2732e8822c620c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-25 06:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f83266c60b795194aff716f1fbb8a528e09e591a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 06:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308afc56046a40c9b8a2191737e50a89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152c9e07afcc4695b0f2c4a4bf896466 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"18edc9c29973c9a7cf35645ddcd2ca356b742e88","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-25 12:41:12.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4d2f9dd9dc7db18fb145675a368752d0282e8e3f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-25 12:53:08.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"102707c12955af330c271015b15cbb962da0f6c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-25 13:39:58.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a1bb1286f5301c784a1dd6028ed5dd3ad6b80fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-25 14:04:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"591d9ac3c6d8bcc5ddc09869a4e5f0b844d1348b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-30 07:37:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a4f0e3ecc80cbdb8bff3f8971b3d8f79e1b4dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 07:41:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a0fd30799454ecd99d3535a8ac11916\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/551d84c5870f40d3888a14d05ff627ca : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"95e93241e1f39cc31da64bf124a61a65d6964fe5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-30 15:45:39.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a973d67dfc4e58da69bb832c906783bbc98aa5ef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-09-30 16:10:42.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c2125cdb93b5c8f8b1ec97b6da4e62cd8a2de71a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-10-10 15:41:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"359baaeacd17ac585bd342b4b5d5b74ecacf862b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 16:01:06.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3249afacf6ead2889418110091ded365c7478881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 18:26:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d129887032360605b8cb2fae4db8a4c2eac78f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:29:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d771f3847fb9453a85dc4ef643408ab5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8558c4e2da6942b1b5e20a6bdc8de641 : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":3},{"id":"ba7aaba1261ca3fd63f8c2f40f2a3dbd459a24db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 18:48:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e52f9fb76493afe3e4cdd6c146fdc46003ad1011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:57:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f84573e89e754ec1a2f5b2f9de6f004a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4a58709e18410d8c0ebfaf62e3d564 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":4},{"id":"831b4ea48fc2bdac5549989820d4a8895184444c","author":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"date":"2025-10-15 09:20:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3228817c4f309c1c8184b4bf6adb63049c88c467","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-10-15 09:32:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e738c23ce669f26e6cd5a6a68f8efc5643b300a9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-29 05:23:00.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c8097c57291a220eba0fd9305d6be0c212cb3893","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-29 15:29:38.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cbd80cde97ea3a2717a30174840bed208f5771ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-05 13:21:22.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47f96c3259e0f75aa6b0d950d5bcbb8bc2ba7692","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-24 06:30:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9c0a1fbc3054821db9a9fe56a027e8b19be6a0f","revisions":{"2a1130eb3df99be75c086241223449655c2af9e6":{"kind":"REWORK","_number":1,"created":"2025-09-25 06:15:28.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/962245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/45/962245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/1"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-09-25 06:14:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-09-25 06:14:58.000000000","tz":0},"subject":"Add new container-registry import method","message":"Add new container-registry import method\n\nChange-Id: I4fe29438a1cae8e705caa1337026c970b564ac72\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/2a1130eb3df99be75c086241223449655c2af9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/2a1130eb3df99be75c086241223449655c2af9e6"}]},"branch":"refs/heads/master"},"9213cc9939b0a147b7cdf5045eec7a6321fe2654":{"kind":"REWORK","_number":2,"created":"2025-09-30 07:37:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/962245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/45/962245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/2"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-09-25 06:14:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-09-30 07:36:48.000000000","tz":0},"subject":"Add new container-registry import method","message":"Add new container-registry import method\n\nChange-Id: I4fe29438a1cae8e705caa1337026c970b564ac72\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9213cc9939b0a147b7cdf5045eec7a6321fe2654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/9213cc9939b0a147b7cdf5045eec7a6321fe2654"}]},"branch":"refs/heads/master"},"6ff7a8d889ef46b0bd3ee2b986492ef32e1ab33d":{"kind":"REWORK","_number":3,"created":"2025-10-10 18:26:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/962245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/45/962245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/3"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-09-25 06:14:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-10 18:24:45.000000000","tz":0},"subject":"Add new container-registry import method","message":"Add new container-registry import method\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: I4fe29438a1cae8e705caa1337026c970b564ac72\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6ff7a8d889ef46b0bd3ee2b986492ef32e1ab33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/6ff7a8d889ef46b0bd3ee2b986492ef32e1ab33d"}]},"branch":"refs/heads/master"},"b9c0a1fbc3054821db9a9fe56a027e8b19be6a0f":{"kind":"REWORK","_number":4,"created":"2025-10-10 18:48:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/45/962245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/45/962245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/45/962245/4"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-09-25 06:14:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-10 18:47:31.000000000","tz":0},"subject":"Add new container-registry import method","message":"Add new container-registry import method\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: I4fe29438a1cae8e705caa1337026c970b564ac72\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/b9c0a1fbc3054821db9a9fe56a027e8b19be6a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/b9c0a1fbc3054821db9a9fe56a027e8b19be6a0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
