)]}'
{"id":"openstack%2Fglance-specs~963239","triplet_id":"openstack%2Fglance-specs~master~Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d","project":"openstack/glance-specs","branch":"master","topic":"bp/download-from-specific-store","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-12-02 15:39:31.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-12-04 14:00:32.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-12-02 17:50:26.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-12-04 13:51:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d","subject":"[spec] Download image from suggested stores","status":"MERGED","created":"2025-10-07 07:05:22.000000000","updated":"2025-12-04 14:01:24.000000000","submitted":"2025-12-04 14:00:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":68,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"963239-bp/download-from-specific-store","meta_rev_id":"92bb8addab8f48fc86aabe0bce7dfaaa16dcfed8","_number":963239,"virtual_id_number":963239,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 14:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-12-02 18:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-12-02 17:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-12-04 13:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2025-12-04 13:51:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 07:13:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-07 16:04:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-10-10 14:11:49.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2025-11-25 14:48:44.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-12-01 14:31:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-02 17:50:26.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"1e27c7bbbf8ec6b1c4f619ebcf6ea5a10e0603be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-07 07:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5f9935d00fc31860a70452b56b1a7ce4e031552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 07:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094a6c33899a47d8845a56beed5a309b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8552fd830087436e8a2421ea98bf53fe : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"dc8d0489cee7d7da6a47b1288eaeb9dffbf241fc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-10-07 16:04:15.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"18ec77312fd0d95c8a6f4b4b7f11514c67ee1d0a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-10-10 14:11:49.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e59bd2b180a83d5233364206e1d793f00abde01e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 14:29:08.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ebaddee21f290d22c42707b99ea18888c27f3378","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-10-10 14:42:13.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c77b0ec6d327ca2a9a15a4eb786e27c893ea04b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 14:52:54.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"60a83ebdd0d4a8c36a2abb038a9abc48b3785f7e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-10-10 15:08:21.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8a26a2b49e4da31d8fcd08cb6eb9449b1f93d350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-10 18:45:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6704cdf7d43b436e85900807974b976503164372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 18:49:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4253c9de08e4edd930269d1bd1644ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785dab0a5d55407db256aa0dbb8479d0 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3087951f8bba800c48e8a2a35ec71442bdd40d70","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-10-24 13:38:17.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"89c83f89a8ec9d06eb9ece20da2e690a1686e25d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-20 07:01:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2cb839b15b8302c5b8f1ff58a4f0b9171c37401e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 07:11:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8702023cdc4136ae57303321080de6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a8322b08014518abf9ff1c4ed5c0b5 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"fc180a819fcb768ab8b26d4962666a51e6662ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-20 07:14:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7bafc1fc7e21791ae37bfde5b25e0877c6fe6e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 07:24:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1822a2500fe643c899817717839a657f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdadf9180bf4f05b83a951690eaeab7 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"1dbe35101f8f0b05a98bc2e7e062e30f1db87991","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-20 14:52:12.000000000","message":"Topic set to bp//download-from-specific-store","accounts_in_message":[],"_revision_number":4},{"id":"2564fded826f68c707a438aedcbecea301e0b6ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 06:44:26.000000000","message":"Topic bp//download-from-specific-store removed","accounts_in_message":[],"_revision_number":4},{"id":"91208f13cdf79093b61170cf4a08aa060558dc60","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-24 06:44:32.000000000","message":"Topic set to bp/download-from-specific-store","accounts_in_message":[],"_revision_number":4},{"id":"9c2d5b9da14a1db263a8c65ccd7dfcec3d160651","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-25 14:24:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"05785ab868fcdd95652258b42304e15cd816da1e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-25 14:48:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"447e69aa9f1125199430e90b22b8936ac1cf67b1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-25 14:53:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"72caf77c5f76087ea5c608411d880a5e293bc35a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-11-25 15:40:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e36103d2d541ae17a0519e0fd3937e66a1a5f039","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-25 16:58:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dda354348624a22f99461f7d841e539ad735b4a5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-11-25 17:01:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a00536631c1bd3f9baf1d8c88407ebe7a5dcd7ce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-11-25 17:04:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1bcbb391ab1f43ea365804cf2809d8c49f54a621","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-11-25 17:26:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e08ee35320062aa0bec838ede2ec35fb4449671","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-01 14:31:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bca6e9b079854dc8ef517ab9788d638668133ba2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-01 14:50:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5779559aa75bd6598684f5d6d561c85ae33a9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-01 14:50:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ff6b3ac147f15f812dc9ae0386476621a7126124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 14:58:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc83b5cfcc44d909da2dd4aafed3ea9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2773aab6b58c4102ba39df4098483df6 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":5},{"id":"165f3109fcf945fa8a5e53fa26d0672435a42ee1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-01 15:02:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"236e47d5ee61c25ee1b393657921b7f78c264acd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-01 15:51:08.000000000","message":"Patch Set 5:\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b2caea895776a556a979c5e46e7721ffae66c55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-02 09:34:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"97517cbb96981f20e1b5128a53a68270e10b3f1f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-02 09:34:48.000000000","message":"Patch Set 5:\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"80e34438975e0c4b4b59af22b7b68b2487168200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 09:37:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd4428d6d8a41cf8bc50b86beebcef9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be084f7473764010aa4c4b75f7f93cca : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":6},{"id":"a9f10169882d8b3a06ba965169bd6bb0f5657515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-02 09:38:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e1576193d630baec2fa7a9f7464216fbd7f5721c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 09:42:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34bab9f634e54468a88c3bc1ce126e90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ce86545bf34f7eb1127df188e1bae4 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":7},{"id":"ec7e0dbce744f388ba89a7a71b31503a95d91091","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-02 13:48:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cd40fbc594746ee554cea200278b12decd577e7d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-12-02 14:35:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"11466aabd0efd754de0c1bb10f3bc5a775b7db14","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-02 15:39:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ba9eaa08167f15afa22fb17cee1494d6d55fbc01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-02 17:16:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"55395d89498a10ce32d4f8181769e05067e1abcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 17:25:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d202f50c3ab5452dab24a8961ee45779\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf9f134be544833834636c268054039 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":8},{"id":"2c5902a63fe11d8c98223647ea0f1cec41ca320d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-12-02 17:50:26.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3cd3e4910b17b0ef14c1ca1d11e07df2ea83a334","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-02 18:15:06.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"523190bed73b93f0ac88a4c3a98ff9aa73a879d1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-04 13:51:38.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"a172de713471de0f359c1efdb7007dfe104e79e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 13:52:01.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0e5c0ee3795ad4bdba7828f82104d85f8cf021b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:00:32.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2be0036c88145a5929c4a5ec93ea94f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4ced6a221e4b07a6c7399559fb6100 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":8},{"id":"c67ef442adfccfcec6e218f6740a80cf146d594d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:00:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"92bb8addab8f48fc86aabe0bce7dfaaa16dcfed8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:01:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63265feb3b504a1a84ecfcf55c14e4c1\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ea49ab7c4e5040ff9955feb847da2d7a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"7ef5651cd1f9fbcbf7cc59530eacea3a5a0110e8","revisions":{"a908cdd7ce7d78d339db6533e90f2f07528dea09":{"kind":"REWORK","_number":1,"created":"2025-10-07 07:05:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/1"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"subject":"[spec] Download image from specific store","message":"[spec] Download image from specific store\n\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a908cdd7ce7d78d339db6533e90f2f07528dea09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/a908cdd7ce7d78d339db6533e90f2f07528dea09"}]},"branch":"refs/heads/master"},"ddb78f7a75d412e03cdbe725565dbdfe0bcbc2c7":{"kind":"REWORK","_number":2,"created":"2025-10-10 18:45:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/2"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-10 18:44:02.000000000","tz":0},"subject":"[spec] Download image from specific store","message":"[spec] Download image from specific store\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/ddb78f7a75d412e03cdbe725565dbdfe0bcbc2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/ddb78f7a75d412e03cdbe725565dbdfe0bcbc2c7"}]},"branch":"refs/heads/master"},"387bbe5b43937c86394986ed87a06942d4366dc8":{"kind":"REWORK","_number":3,"created":"2025-11-20 07:01:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/3"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-20 06:59:12.000000000","tz":0},"subject":"[spec] Download image from specific store","message":"[spec] Download image from specific store\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/387bbe5b43937c86394986ed87a06942d4366dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/387bbe5b43937c86394986ed87a06942d4366dc8"}]},"branch":"refs/heads/master"},"de9afb829fce71a588c4670ec7b9a965620b89f4":{"kind":"REWORK","_number":4,"created":"2025-11-20 07:14:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/4"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-20 07:11:53.000000000","tz":0},"subject":"[spec] Download image from specific store","message":"[spec] Download image from specific store\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/de9afb829fce71a588c4670ec7b9a965620b89f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/de9afb829fce71a588c4670ec7b9a965620b89f4"}]},"branch":"refs/heads/master"},"7a723233bb79d1db0af4fed5366178930e55764f":{"kind":"REWORK","_number":5,"created":"2025-12-01 14:50:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/5"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-01 14:50:06.000000000","tz":0},"subject":"[spec] Download image from specific store","message":"[spec] Download image from specific store\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/7a723233bb79d1db0af4fed5366178930e55764f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/7a723233bb79d1db0af4fed5366178930e55764f"}]},"branch":"refs/heads/master"},"585313a0df0fec5694bb0ebc78530c530d4ba1bd":{"kind":"REWORK","_number":6,"created":"2025-12-02 09:34:12.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/6"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-02 09:33:52.000000000","tz":0},"subject":"[spec] Download image from specific store","message":"[spec] Download image from specific store\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/585313a0df0fec5694bb0ebc78530c530d4ba1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/585313a0df0fec5694bb0ebc78530c530d4ba1bd"}]},"branch":"refs/heads/master"},"c14b5ccd3b5e22a6cdac09b3ba7760c9521be013":{"kind":"REWORK","_number":7,"created":"2025-12-02 09:38:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/7"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-02 09:37:12.000000000","tz":0},"subject":"[spec] Download image from suggested stores","message":"[spec] Download image from suggested stores\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/c14b5ccd3b5e22a6cdac09b3ba7760c9521be013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/c14b5ccd3b5e22a6cdac09b3ba7760c9521be013"}]},"branch":"refs/heads/master"},"7ef5651cd1f9fbcbf7cc59530eacea3a5a0110e8":{"kind":"REWORK","_number":8,"created":"2025-12-02 17:16:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/963239/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/39/963239/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/39/963239/8"}}},"commit":{"parents":[{"commit":"d15bbe1290fbef90d968394fbd6deb15d87b30a4","subject":"Add 2026.1 directory for specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/d15bbe1290fbef90d968394fbd6deb15d87b30a4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-10-07 06:54:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-02 17:16:06.000000000","tz":0},"subject":"[spec] Download image from suggested stores","message":"[spec] Download image from suggested stores\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Ib9541a406b98eb0ceb181e2c0825f86c9ca7bc2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/7ef5651cd1f9fbcbf7cc59530eacea3a5a0110e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/7ef5651cd1f9fbcbf7cc59530eacea3a5a0110e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
