)]}'
{"id":"openstack%2Fglance-specs~984902","triplet_id":"openstack%2Fglance-specs~master~Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65","project":"openstack/glance-specs","branch":"master","topic":"bp/add-volume-backup-based-image","attention_set":{"36712":{"account":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"last_update":"2026-04-16 07:46:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65","subject":"[Spec] Create volume backup based Image","status":"NEW","created":"2026-04-16 07:43:13.000000000","updated":"2026-04-16 09:50:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"448ec0883d269db40f40067dec3f64d78ba7198d","_number":984902,"virtual_id_number":984902,"owner":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-16 09:50:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 07:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"223cbc075adf9e2a1bb47e7d3a160502fc16f2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"date":"2026-04-16 07:43:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dee49a928231a4a637e18dd5138583620dd9779c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 07:46:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f7a87e844f94f05b39b2d34e6bd42b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a76236404343128f91c06d38a75e55 : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ff4cba6adc485aa768543ed5d2bc8f : FAILURE in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a8d155eb93b6d5cdff3573979e599b0d40b90a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"date":"2026-04-16 08:01:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6f0c931d58d97bde5c298ca93c04b2423a8a2350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"date":"2026-04-16 08:08:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dfa81fbaac6cb4abe019f1b08d8716c9e855f9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 08:12:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5cafcb93dcb417fa767ab893b567846\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40bd2fe1b6014aa7b06f2b1000ab46d7 : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70da3574d8e14194ad1cd24dcb76d7a9 : FAILURE in 2m 15s","accounts_in_message":[],"_revision_number":3},{"id":"4f9d439f2b6d1f5d9191fdfd17483c331bc6729f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"date":"2026-04-16 09:25:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9be82819ce05258793e751ea864e93a21ef8a9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 09:34:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/110713a7a5354d3ab6e78ea1cbd2d2b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070773bbf070461d8b44158758701ae3 : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71dd01b9e4e84c419acc2f2df226fed7 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":4},{"id":"49a480914b88015376dacf764b2c8a8a88c3457f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"date":"2026-04-16 09:41:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"448ec0883d269db40f40067dec3f64d78ba7198d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 09:50:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2229a274368d4fc1a9f56369d8ffc717\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0569b8c49d5e441caf2233f9050d6407 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88d3665b9a44ff0af612bccc0d541e1 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"87c6a0c43c071be880b08b945cbf8b5c264b744c","revisions":{"1fdecba9692a82a386d4e9ee07d626d518d33979":{"kind":"REWORK","_number":1,"created":"2026-04-16 07:43:13.000000000","uploader":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"ref":"refs/changes/02/984902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/02/984902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/1"}}},"commit":{"parents":[{"commit":"398409005007df3ab1126d690a09eceaf9e7011e","subject":"Merge \"Add 2026.2 directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/398409005007df3ab1126d690a09eceaf9e7011e"}]}],"author":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-04-15 07:16:54.000000000","tz":540},"committer":{"name":"yunokim","email":"yunho01.kim@samsung.com","date":"2026-04-16 07:42:41.000000000","tz":540},"subject":"[Spec] Create volume backup based Image","message":"[Spec] Create volume backup based Image\n\nThis spec propose to create an image based on volume backups.\nThis enables the creation and utilization if images for multi-volume\nVMs, without the constraints of backend storage.\n\nBlueprint: https://blueprints.launchpad.net/glance/+spec/add-volume-backup-based-image\nChange-Id: Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65\nSigned-off-by: Yunho Kim \u003cyunho01.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/1fdecba9692a82a386d4e9ee07d626d518d33979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/1fdecba9692a82a386d4e9ee07d626d518d33979"}]},"branch":"refs/heads/master"},"1758a29f1897926c287c138fc4e0c2e98151f835":{"kind":"REWORK","_number":2,"created":"2026-04-16 08:01:08.000000000","uploader":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"ref":"refs/changes/02/984902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/02/984902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/2"}}},"commit":{"parents":[{"commit":"398409005007df3ab1126d690a09eceaf9e7011e","subject":"Merge \"Add 2026.2 directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/398409005007df3ab1126d690a09eceaf9e7011e"}]}],"author":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-04-15 07:16:54.000000000","tz":540},"committer":{"name":"yunokim","email":"yunho01.kim@samsung.com","date":"2026-04-16 08:01:10.000000000","tz":540},"subject":"[Spec] Create volume backup based Image","message":"[Spec] Create volume backup based Image\n\nThis spec propose to create an image based on volume backups.\nThis enables the creation and utilization if images for multi-volume\nVMs, without the constraints of backend storage.\n\nBlueprint: https://blueprints.launchpad.net/glance/+spec/add-volume-backup-based-image\nChange-Id: Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65\nSigned-off-by: Yunho Kim \u003cyunho01.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/1758a29f1897926c287c138fc4e0c2e98151f835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/1758a29f1897926c287c138fc4e0c2e98151f835"}]},"branch":"refs/heads/master"},"e6129e51284f574ad9e47978baa90e314a4306f1":{"kind":"REWORK","_number":3,"created":"2026-04-16 08:08:11.000000000","uploader":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"ref":"refs/changes/02/984902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/02/984902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/3"}}},"commit":{"parents":[{"commit":"398409005007df3ab1126d690a09eceaf9e7011e","subject":"Merge \"Add 2026.2 directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/398409005007df3ab1126d690a09eceaf9e7011e"}]}],"author":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-04-15 07:16:54.000000000","tz":540},"committer":{"name":"yunokim","email":"yunho01.kim@samsung.com","date":"2026-04-16 08:08:12.000000000","tz":540},"subject":"[Spec] Create volume backup based Image","message":"[Spec] Create volume backup based Image\n\nThis spec propose to create an image based on volume backups.\nThis enables the creation and utilization if images for multi-volume\nVMs, without the constraints of backend storage.\n\nBlueprint: https://blueprints.launchpad.net/glance/+spec/add-volume-backup-based-image\nChange-Id: Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65\nSigned-off-by: Yunho Kim \u003cyunho01.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e6129e51284f574ad9e47978baa90e314a4306f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/e6129e51284f574ad9e47978baa90e314a4306f1"}]},"branch":"refs/heads/master"},"3dffd66ce36315f296a71c51ead998e78bfb4a30":{"kind":"REWORK","_number":4,"created":"2026-04-16 09:25:36.000000000","uploader":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"ref":"refs/changes/02/984902/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/02/984902/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/4"}}},"commit":{"parents":[{"commit":"398409005007df3ab1126d690a09eceaf9e7011e","subject":"Merge \"Add 2026.2 directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/398409005007df3ab1126d690a09eceaf9e7011e"}]}],"author":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-04-15 07:16:54.000000000","tz":540},"committer":{"name":"yunokim","email":"yunho01.kim@samsung.com","date":"2026-04-16 09:25:40.000000000","tz":540},"subject":"[Spec] Create volume backup based Image","message":"[Spec] Create volume backup based Image\n\nThis spec propose to create an image based on volume backups.\nThis enables the creation and utilization if images for multi-volume\nVMs, without the constraints of backend storage.\n\nBlueprint: https://blueprints.launchpad.net/glance/+spec/add-volume-backup-based-image\nChange-Id: Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65\nSigned-off-by: Yunho Kim \u003cyunho01.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/3dffd66ce36315f296a71c51ead998e78bfb4a30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/3dffd66ce36315f296a71c51ead998e78bfb4a30"}]},"branch":"refs/heads/master"},"87c6a0c43c071be880b08b945cbf8b5c264b744c":{"kind":"REWORK","_number":5,"created":"2026-04-16 09:41:49.000000000","uploader":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"ref":"refs/changes/02/984902/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-specs","ref":"refs/changes/02/984902/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-specs refs/changes/02/984902/5"}}},"commit":{"parents":[{"commit":"398409005007df3ab1126d690a09eceaf9e7011e","subject":"Merge \"Add 2026.2 directory for specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/398409005007df3ab1126d690a09eceaf9e7011e"}]}],"author":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-04-15 07:16:54.000000000","tz":540},"committer":{"name":"yunokim","email":"yunho01.kim@samsung.com","date":"2026-04-16 09:41:52.000000000","tz":540},"subject":"[Spec] Create volume backup based Image","message":"[Spec] Create volume backup based Image\n\nThis spec propose to create an image based on volume backups.\nThis enables the creation and utilization if images for multi-volume\nVMs, without the constraints of backend storage.\n\nBlueprint: https://blueprints.launchpad.net/glance/+spec/add-volume-backup-based-image\nChange-Id: Id3e1cb8b4fa31fa55839d1c9438c5752ae1c2c65\nSigned-off-by: Yunho Kim \u003cyunho01.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/87c6a0c43c071be880b08b945cbf8b5c264b744c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-specs/commit/87c6a0c43c071be880b08b945cbf8b5c264b744c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
