)]}'
{"id":"openstack%2Fglance-tempest-plugin~822324","triplet_id":"openstack%2Fglance-tempest-plugin~master~I451e21795c79815d791530b70a3c4936f373be33","project":"openstack/glance-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2025-04-14 14:07:29.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-04-14 14:07:29.000000000","reason":"Change was abandoned"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2025-04-14 14:07:29.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-04-28 06:15:53.000000000","reason":"Abhishek Kekane replied on the change"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-04-14 14:07:29.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-04-14 14:07:29.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-04-14 14:07:29.000000000","reason":"Change was abandoned"},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2025-04-07 10:13:39.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}}},"hashtags":[],"change_id":"I451e21795c79815d791530b70a3c4936f373be33","subject":"Add tests to support new manager role","status":"ABANDONED","created":"2021-12-20 13:48:39.000000000","updated":"2025-04-14 14:07:29.000000000","total_comment_count":15,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"5996d8251f2439bc443f57df210db8969375e02f","_number":822324,"virtual_id_number":822324,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-03 09:05:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 14:17:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-12-20 14:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-23 06:00:16.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-12-23 06:00:16.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-12-23 06:00:16.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-12-23 06:00:16.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-12-23 06:00:16.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2022-04-29 08:39:35.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-05-05 19:14:22.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"state":"CC"},{"updated":"2025-04-07 10:13:39.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"CC"}],"messages":[{"id":"9a9cbb4c89a090c1e9e8329d175332a14af28017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-20 13:48:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89025b631f32492b604659f4d35ba32fe4230927","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-12-20 14:17:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6f55adb412d4378e1bac8ad4e6da46987f40443b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-12-20 14:19:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a613501f52f389405178025baed1e3b8c3a700b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:31:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a553213ff014e63a4925813016078ff : SUCCESS in 5m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6915ea85a6d54db9bbeb413bc4502146 : SUCCESS in 35m 47s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0408ae8efca64650b7571e024912e204 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c9c7daa0bca834eba5778156936adddd47051dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-21 10:44:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08c10a6615df2fbaa6ce65ae33060652c130af2e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-21 10:45:25.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"953915de3ab74a43e828c87623b1ff1e900b01ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 11:20:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f887f34316e4275865f95372da49514 : SUCCESS in 3m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f7135a5991843dcbf57aeb73a94ce9f : FAILURE in 32m 37s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3ec9b2b4a53149ca8de606fa651d6a50 : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":2},{"id":"153145749040e9e8ddcae8421770846865942d50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-22 08:55:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"64077dc26330c232448e2deb05a9f95befa50d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 09:17:39.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"a0561854835c1b85991536693faf66c4f2c37063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-22 10:56:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"16630ca0d411247c564f56d2f2c097abd29580eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 11:27:25.000000000","message":"Patch Set 4: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c5e1264acd4651bcca85030e4c5bac : FAILURE in 4m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6883295b40244f03865ed19ebbe718ed : SUCCESS in 23m 02s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7849eaf596845669c2e7b41f563295d : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":4},{"id":"40fadd1ca6dcd66dc7df177c2909842cd9eab2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-12-22 12:10:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c1ba3e87ae8a73428c494564ce73957aa5c9737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:57:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026fca7dec0946bd8436c26f490f29ac : SUCCESS in 3m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e94be61328884a85b8e6ad15e6b73ec2 : SUCCESS in 35m 52s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8489c2ec354e4acf9e14793a7818d49b : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":5},{"id":"80f42ef7fc1f9876ddfe0571c6fa1c5cfadddd1d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-01-13 01:27:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"30933e1e567340b85f0300c25181944b3d0d244b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-04-28 06:15:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6debb2132a3a6af1acbdae8948e07dcb90b62da0","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-04-29 08:39:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b49751ff10ffb1bf1d2a70a41f34df0ed9162826","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-05-05 19:14:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a605f7de3acd893b472ba2f3209a585109bb344f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-07-06 13:56:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7dbba928cbd51a70128772dd7da0ee6007a95fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 14:36:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/641bd15df6dd4068875a5a54f5d3e5aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e25b49f433643af9c7f1960958e1b03 : SUCCESS in 3m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f62568d03da64308a530df2ae25ddb68 : FAILURE in 33m 52s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fbb26bd2df948b58e73cfc63a12b517 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":6},{"id":"b18f25126dec77fc853f89c6193a88c03cb95a8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-04-03 09:05:26.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"a4343899764d8a3f4d2efc1c8141d47fed32c54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 09:05:53.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tempest for 822318,5","accounts_in_message":[],"_revision_number":7},{"id":"4fa893ba2702bd409576b41fa23e71dfa13e5da3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-03 14:06:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d855ed3b44292501d62dab0e148a55e6856364c","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-04-07 10:13:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5996d8251f2439bc443f57df210db8969375e02f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-04-14 14:07:29.000000000","message":"Abandoned\n\nDuring PTG, we decided to abandon this","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6bb9172c64ecba2140acdebe5475946fa3619411","revisions":{"50506e3b7d8734647a05af29ad6c70b2b60e197d":{"kind":"REWORK","_number":1,"created":"2021-12-20 13:48:39.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/822324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/1"}}},"commit":{"parents":[{"commit":"c9bfadc59b8b2954d69de494e09fc3429a0fd8b9","subject":"Implement API protection testing for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c9bfadc59b8b2954d69de494e09fc3429a0fd8b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 13:43:50.000000000","tz":0},"subject":"Verify new manager persona supported for image apis","message":"Verify new manager persona supported for image apis\n\nSince the introduction of the base capability has been extended\nas a result of an OpenStack community goal to include a ``manager``\nrole in the project scope.\n\nThis commit verifies whether all the image APIs tempest tests\nwhich are under ``member`` role are supported with \u0027manager\u0027\nrole or not.\n\nDepends-On: Ida889aa30d462443b801c0f524c51f54b8b756d5\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/50506e3b7d8734647a05af29ad6c70b2b60e197d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/50506e3b7d8734647a05af29ad6c70b2b60e197d"}]},"branch":"refs/heads/master"},"5184afe9dfc6501178721ae7f53e2a60b591e508":{"kind":"REWORK","_number":2,"created":"2021-12-21 10:44:42.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/822324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/2"}}},"commit":{"parents":[{"commit":"c9bfadc59b8b2954d69de494e09fc3429a0fd8b9","subject":"Implement API protection testing for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c9bfadc59b8b2954d69de494e09fc3429a0fd8b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-21 10:33:01.000000000","tz":0},"subject":"Add tests to support new manager role","message":"Add tests to support new manager role\n\nDuring Yoga PTG community has decided to add one more\n`manager` role/persona [1] which will have more privileges\nthan `member` role and less than `admin` role.\n\nThis commit verifies that glance APIs support newly introduced\n`manager` role/persona.\n\n[1]: https://review.opendev.org/c/openstack/governance/+/815158\n\nDepends-On: I4a0d82bfeb51da0b9cc6f00346ef576f8dfedc96\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/5184afe9dfc6501178721ae7f53e2a60b591e508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/5184afe9dfc6501178721ae7f53e2a60b591e508"}]},"branch":"refs/heads/master"},"35d7e3d16217443e0358fdaeafad3c684c73fa6e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-12-22 08:55:39.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/822324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/3"}}},"commit":{"parents":[{"commit":"c9bfadc59b8b2954d69de494e09fc3429a0fd8b9","subject":"Implement API protection testing for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c9bfadc59b8b2954d69de494e09fc3429a0fd8b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-22 08:55:38.000000000","tz":0},"subject":"Add tests to support new manager role","message":"Add tests to support new manager role\n\nDuring Yoga PTG community has decided to add one more\n`manager` role/persona [1] which will have more privileges\nthan `member` role and less than `admin` role.\n\nThis commit verifies that glance APIs support newly introduced\n`manager` role/persona.\n\n[1]: https://review.opendev.org/c/openstack/governance/+/815158\n\nDepends-On: I4149e99bdc50194dd4067f5f3e2f3019e48b5a59\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/35d7e3d16217443e0358fdaeafad3c684c73fa6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/35d7e3d16217443e0358fdaeafad3c684c73fa6e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"896edc0ccc18d1aed81a9b24e2955788d81e7fd9":{"kind":"REWORK","_number":4,"created":"2021-12-22 10:56:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/822324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/4"}}},"commit":{"parents":[{"commit":"c9bfadc59b8b2954d69de494e09fc3429a0fd8b9","subject":"Implement API protection testing for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c9bfadc59b8b2954d69de494e09fc3429a0fd8b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-22 10:49:42.000000000","tz":0},"subject":"Add tests to support new manager role","message":"Add tests to support new manager role\n\nDuring Yoga PTG community has decided to add one more\n`manager` role/persona [1] which will have more privileges\nthan `member` role and less than `admin` role.\n\nThis commit verifies that glance APIs support newly introduced\n`manager` role/persona.\n\n[1]: https://review.opendev.org/c/openstack/governance/+/815158\n\nDepends-On: I4149e99bdc50194dd4067f5f3e2f3019e48b5a59\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/896edc0ccc18d1aed81a9b24e2955788d81e7fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/896edc0ccc18d1aed81a9b24e2955788d81e7fd9"}]},"branch":"refs/heads/master"},"614d6bf88afe6e70853ec8b28fc7baffa7dd5556":{"kind":"REWORK","_number":5,"created":"2021-12-22 12:10:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/822324/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/5"}}},"commit":{"parents":[{"commit":"c9bfadc59b8b2954d69de494e09fc3429a0fd8b9","subject":"Implement API protection testing for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c9bfadc59b8b2954d69de494e09fc3429a0fd8b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-22 12:05:27.000000000","tz":0},"subject":"Add tests to support new manager role","message":"Add tests to support new manager role\n\nDuring Yoga PTG community has decided to add one more\n`manager` role/persona [1] which will have more privileges\nthan `member` role and less than `admin` role.\n\nThis commit verifies that glance APIs support newly introduced\n`manager` role/persona.\n\n[1]: https://review.opendev.org/c/openstack/governance/+/815158\n\nDepends-On: I4149e99bdc50194dd4067f5f3e2f3019e48b5a59\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/614d6bf88afe6e70853ec8b28fc7baffa7dd5556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/614d6bf88afe6e70853ec8b28fc7baffa7dd5556"}]},"branch":"refs/heads/master"},"e45f31cd4c57633aad86c86e3a2f10f40959e833":{"kind":"REWORK","_number":6,"created":"2022-07-06 13:56:14.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/24/822324/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/6"}}},"commit":{"parents":[{"commit":"efdd6972f21b14e531ba0f42f2b32def12dac940","subject":"Refactor rbac metadef tests structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/efdd6972f21b14e531ba0f42f2b32def12dac940"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2022-07-06 13:53:43.000000000","tz":330},"subject":"Add tests to support new manager role","message":"Add tests to support new manager role\n\nDuring Yoga PTG community has decided to add one more\n`manager` role/persona [1] which will have more privileges\nthan `member` role and less than `admin` role.\n\nThis commit verifies that glance APIs support newly introduced\n`manager` role/persona.\n\n[1]: https://review.opendev.org/c/openstack/governance/+/815158\n\nDepends-On: I4149e99bdc50194dd4067f5f3e2f3019e48b5a59\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/e45f31cd4c57633aad86c86e3a2f10f40959e833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/e45f31cd4c57633aad86c86e3a2f10f40959e833"}]},"branch":"refs/heads/master"},"6bb9172c64ecba2140acdebe5475946fa3619411":{"kind":"REWORK","_number":7,"created":"2025-04-03 09:05:26.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/24/822324/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/24/822324/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/24/822324/7"}}},"commit":{"parents":[{"commit":"fd2087db9ac50706d396324f1f34da483e343e15","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/fd2087db9ac50706d396324f1f34da483e343e15"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-12-20 12:49:06.000000000","tz":0},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-04-03 08:54:18.000000000","tz":180},"subject":"Add tests to support new manager role","message":"Add tests to support new manager role\n\nDuring Yoga PTG community has decided to add one more\n`manager` role/persona [1] which will have more privileges\nthan `member` role and less than `admin` role.\n\nThis commit verifies that glance APIs support newly introduced\n`manager` role/persona.\n\n[1]: https://review.opendev.org/c/openstack/governance/+/815158\n\nDepends-On: I4149e99bdc50194dd4067f5f3e2f3019e48b5a59\nDepends-On: I29a14118126efd3bfb168b25df36911433cad2d9\nChange-Id: I451e21795c79815d791530b70a3c4936f373be33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/6bb9172c64ecba2140acdebe5475946fa3619411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/6bb9172c64ecba2140acdebe5475946fa3619411"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
