)]}'
{"id":"openstack%2Fglance-tempest-plugin~882566","triplet_id":"openstack%2Fglance-tempest-plugin~master~I7d5fa2f9b761d9466ed1534d12d87d19aa9644bb","project":"openstack/glance-tempest-plugin","branch":"master","topic":"no-devstack-gate","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-09-23 14:27:28.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-09-23 14:27:28.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-23 14:25:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-09-23 14:27:28.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-09-23 14:27:28.000000000","reason":"Change was abandoned"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2024-09-23 14:27:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7d5fa2f9b761d9466ed1534d12d87d19aa9644bb","subject":"zuul: remove devstack-gate dependency","status":"ABANDONED","created":"2023-05-08 14:44:17.000000000","updated":"2024-09-23 14:27:28.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"069c7d61a46e4d04ffd49c245c412ca0c0a6dab7","_number":882566,"virtual_id_number":882566,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2024-09-23 14:25:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-09-23 14:16:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2024-09-23 14:25:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2024-09-23 14:25:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 14:46:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-05-08 15:16:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 12:23:30.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-01-25 13:07:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-17 17:49:26.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-09-17 17:49:26.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-09-23 14:16:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"85f4699fd29e6d8e75f96482793fe6713140490f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-08 14:44:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66c91b3fdbac4ddeb62cbf41d23ce5527278fee2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-05-08 14:46:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"280c45d053120a19e4064d90cb1e592c3cfafae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 15:16:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33378beba1e240a6bcc26755e1bd0418\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f161313c10ca4deb904ebb52a564fed3 : SUCCESS in 3m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07acd693ba6d464081479c7a410c74eb : SUCCESS in 30m 31s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75552371be344c4dafc3f0c6023eaa0d : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"aced49369664a29a0dd336ae9b131126546f099f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-05-30 12:23:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2225d1d1a39b3820f8a537c70d4264910cb9fd5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 06:08:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"074388fc61bbecdeae74eb7730ce424f2f832abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 06:56:50.000000000","message":"Patch Set 2: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/754d6a2bc3724eaab8a6ac06b6c761b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99026f82d0b4f7d809e063553662317 : SUCCESS in 2m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1670a33f1956426e913916c24a9352d6 : SUCCESS in 24m 49s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f49201edee14555afd72e5acbdfcc30 : SUCCESS in 23m 36s\n- glance-secure-rbac-protection-functional-2023-1 https://zuul.opendev.org/t/openstack/build/9dcec6aeee2e4995befde2fd0c525f48 : SUCCESS in 23m 42s\n- glance-secure-rbac-protection-functional-zed https://zuul.opendev.org/t/openstack/build/5063fe2df18b4bd186dbeb62cbb05a87 : SUCCESS in 30m 20s\n- glance-secure-rbac-protection-functional-yoga https://zuul.opendev.org/t/openstack/build/6b69efc4c7ab4dff9843e192c6c6292e : SUCCESS in 31m 47s\n- glance-secure-rbac-protection-functional-xena https://zuul.opendev.org/t/openstack/build/d4acacdce9c546729646335220416b91 : SUCCESS in 35m 29s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/glance-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"branches\" attribute\nof the \"glance-secure-rbac-protection-functional\" job stanza:\n\n  job:\n      name: glance-secure-rbac-protection-functional\n      parent: devstack-tempest\n      timeout: 4200\n      branches: ^(?!stable/wallaby).*$\n      required-projects:\n        - openstack/glance\n        - openstack/glance-tempest-plugin\n      vars: \u0026rbac-protection-functional-vars\n  ...\n\n  in \"openstack/glance-tempest-plugin/.zuul.yaml@master\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"1324a27fc7869c6ff25d5abe1f0488374b967a63","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 13:07:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5c3e3df27f7b054924cbcaac0ded200da583b6b2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2024-09-23 13:07:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36627efbdee574f12ce9263b7f3f27bb4a2df9a3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-23 14:16:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c25858b2d00c4bd6ac91dbd2c8e18a997c619a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 14:17:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e85b5173ca8d2915027a91ac4c1c0bf4a4be6d5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-23 14:25:40.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"674837bd0ddf63d87576e31421c905d7049b7623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 14:25:51.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"67931c3fb7b6d786090b472f98794ca9aba95367","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-23 14:25:54.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"069c7d61a46e4d04ffd49c245c412ca0c0a6dab7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-09-23 14:27:28.000000000","message":"Abandoned\n\nAlready done as part of https://review.opendev.org/c/openstack/glance-tempest-plugin/+/916332","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7ba3513c81339fa549f62686a4e0cc56c638049","revisions":{"008f3d88edabcf7dfb1139db116be6b2d2f19e5b":{"kind":"REWORK","_number":1,"created":"2023-05-08 14:44:17.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/66/882566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/66/882566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/1"}}},"commit":{"parents":[{"commit":"eec965627283aba555ff593645ff8551be41efd4","subject":"Make rbac functional job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/eec965627283aba555ff593645ff8551be41efd4"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2023-05-08 14:42:32.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2023-05-08 14:44:15.000000000","tz":120},"subject":"zuul: remove devstack-gate dependency","message":"zuul: remove devstack-gate dependency\n\ndevstack-gate is never needed in never zuul v3 jobs.\n\nChange-Id: I7d5fa2f9b761d9466ed1534d12d87d19aa9644bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/008f3d88edabcf7dfb1139db116be6b2d2f19e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/008f3d88edabcf7dfb1139db116be6b2d2f19e5b"}]},"branch":"refs/heads/master"},"c7ba3513c81339fa549f62686a4e0cc56c638049":{"kind":"REWORK","_number":2,"created":"2024-01-22 06:08:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/882566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/66/882566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/66/882566/2"}}},"commit":{"parents":[{"commit":"d6f7287947731da139413899ddddddcfb95e1a50","subject":"glance-tempest-plugin branchless: add stable branch testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/d6f7287947731da139413899ddddddcfb95e1a50"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 06:07:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 06:07:41.000000000","tz":540},"subject":"zuul: remove devstack-gate dependency","message":"zuul: remove devstack-gate dependency\n\ndevstack-gate is never needed in never zuul v3 jobs.\n\nChange-Id: I7d5fa2f9b761d9466ed1534d12d87d19aa9644bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c7ba3513c81339fa549f62686a4e0cc56c638049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c7ba3513c81339fa549f62686a4e0cc56c638049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
