)]}'
{"id":"openstack%2Fglance-tempest-plugin~884006","triplet_id":"openstack%2Fglance-tempest-plugin~master~I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e","project":"openstack/glance-tempest-plugin","branch":"master","topic":"srbac-downstream-failures","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-11-19 15:17:32.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-11-19 15:17:32.000000000","reason":"Change was abandoned"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2023-05-24 08:48:11.000000000","reason":"\u003cGERRIT_ACCOUNT_10459\u003e replied on the change","reason_account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2024-11-19 15:17:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e","subject":"Fix admin client to system admin","status":"ABANDONED","created":"2023-05-24 08:26:11.000000000","updated":"2024-11-19 15:17:32.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c9fa46b41bb727302382cf138103fdbc805c4e44","_number":884006,"virtual_id_number":884006,"owner":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-12 21:20:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 08:27:52.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-05-24 08:27:52.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-05-24 08:36:12.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2023-05-24 09:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 20:42:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"534075b48f4f961993f506878dbb173f9d37c817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2023-05-24 08:26:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8e0cfd7e811c4c66fb15bb20164c2819617ea29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2023-05-24 08:34:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"24f09433d5e764e0bd2eb43b1488bf16e0e90a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2023-05-24 08:36:49.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"66ddd621cf9006b3c0fb57c72566bf56d8275465","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-24 08:48:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20afd9e3ce4d7e1d27c911777a017eecabf5116d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:09:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/275a05e83d67437e8a938b9103c18c11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788fdde2830c44b6b5932e93c1431e8a : SUCCESS in 2m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/02259b56a6c14e8ba9ea49f4d7cf09e8 : FAILURE in 30m 09s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ccc487b83764edd9681e71a6afecc0a : FAILURE in 29m 48s\n- glance-secure-rbac-protection-functional-2023-1 https://zuul.opendev.org/t/openstack/build/694dc435cbf047a08ec3605868e231c7 : FAILURE in 29m 09s\n- glance-secure-rbac-protection-functional-zed https://zuul.opendev.org/t/openstack/build/bfa6bb67c2da46b3999381894836a95e : FAILURE in 30m 35s\n- glance-secure-rbac-protection-functional-yoga https://zuul.opendev.org/t/openstack/build/67f9f7eeaaf14408916b5085fb4cbe92 : SUCCESS in 31m 34s\n- glance-secure-rbac-protection-functional-xena https://zuul.opendev.org/t/openstack/build/3ca1db24e389497d8d3126879fb44454 : FAILURE in 28m 16s","accounts_in_message":[],"_revision_number":3},{"id":"f91b8af722508d4bfb5de84bae077dbd9e13bd2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2023-05-24 09:13:49.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7cc538688e0282ffa4631e25128a6d5a99815a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:58:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b8744bfd74a4ec6999e8ba575563d2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6af43c69b8044b3a07ff561ee4c746f : SUCCESS in 3m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0bbda5a3199043daa1556f6103cfc44d : FAILURE in 34m 52s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2eda810d31f456288a1884c3e04979d : FAILURE in 37m 54s\n- glance-secure-rbac-protection-functional-2023-1 https://zuul.opendev.org/t/openstack/build/992c413be0164bc885de2887532ebc25 : FAILURE in 35m 51s\n- glance-secure-rbac-protection-functional-zed https://zuul.opendev.org/t/openstack/build/eb9e8bcfb26e47d1baaa8e78531cd5d6 : FAILURE in 34m 13s\n- glance-secure-rbac-protection-functional-yoga https://zuul.opendev.org/t/openstack/build/689c24dfff424f30a12d0d117b7c4320 : SUCCESS in 22m 28s\n- glance-secure-rbac-protection-functional-xena https://zuul.opendev.org/t/openstack/build/6211037165f44104b9076c81b13d951d : FAILURE in 23m 10s","accounts_in_message":[],"_revision_number":4},{"id":"4bbe66ed738a9bf175318b4ce016bb28e59812fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"real_author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-12 20:42:07.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_34510\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"_revision_number":5},{"id":"db769fc0977622b6e39d260a190d1bc393c71931","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-12 20:42:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7a1d218479ea0435db1ce81dd32a4bcdedfa1bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 21:20:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b4f0147b1974f26bc0ce632097b86d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44ad4825f1f4d8793a7c2184d077f3b : SUCCESS in 2m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/10bcc56de0484ea4a89976c90de238ab : FAILURE in 21m 33s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08d6278404654ac893f9b70d8bdab2fd : FAILURE in 17m 48s\n- glance-secure-rbac-protection-functional-2024-1 https://zuul.opendev.org/t/openstack/build/cac77a19b3184185a9c9c61f6451ea6d : FAILURE in 32m 36s\n- glance-secure-rbac-protection-functional-2023-2 https://zuul.opendev.org/t/openstack/build/391fbdfa942e48c4985330690f032c74 : FAILURE in 24m 22s\n- glance-secure-rbac-protection-functional-2023-1 https://zuul.opendev.org/t/openstack/build/3e9d225ba10249a191e4903105bc1e55 : FAILURE in 23m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c9fa46b41bb727302382cf138103fdbc805c4e44","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2024-11-19 15:17:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cf1d8a4071ac98d77c7eb93fd30c47106be739bc","revisions":{"cf621d21d134e96509072818c0e506cffa54a88d":{"kind":"REWORK","_number":1,"created":"2023-05-24 08:26:11.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/06/884006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/06/884006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/1"}}},"commit":{"parents":[{"commit":"87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b","subject":"Merge \"Refactor rbac tests structure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b"}]}],"author":{"name":"msava","email":"msava@redhat.com","date":"2023-05-24 08:17:31.000000000","tz":180},"committer":{"name":"msava","email":"msava@redhat.com","date":"2023-05-24 08:26:09.000000000","tz":180},"subject":"Fix admin client to system admin","message":"Fix admin client to system admin\n\nGlance now support system scope.This fix change admin_client to os_system_admin.\nThis fix downstream srbac tests failures.\n\nChange-Id: I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/cf621d21d134e96509072818c0e506cffa54a88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/cf621d21d134e96509072818c0e506cffa54a88d"}]},"branch":"refs/heads/master"},"c6a77038cbc257e3c44dbf0b6c5ca5b019fbb931":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-24 08:34:00.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/06/884006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/06/884006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/2"}}},"commit":{"parents":[{"commit":"87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b","subject":"Merge \"Refactor rbac tests structure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b"}]}],"author":{"name":"msava","email":"msava@redhat.com","date":"2023-05-24 08:17:31.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2023-05-24 08:34:00.000000000","tz":0},"subject":"Fix admin client to system admin","message":"Fix admin client to system admin\n\nGlance now support system scope.This patch change admin_client to os_system_admin and downstream srbac tests failures.\n\nChange-Id: I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c6a77038cbc257e3c44dbf0b6c5ca5b019fbb931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c6a77038cbc257e3c44dbf0b6c5ca5b019fbb931"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e32aa10c85e910b616420a89665743f2a23b78d1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-24 08:36:49.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/06/884006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/06/884006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/3"}}},"commit":{"parents":[{"commit":"87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b","subject":"Merge \"Refactor rbac tests structure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b"}]}],"author":{"name":"msava","email":"msava@redhat.com","date":"2023-05-24 08:17:31.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2023-05-24 08:36:49.000000000","tz":0},"subject":"Fix admin client to system admin","message":"Fix admin client to system admin\n\nGlance now support system scope.This patch fix admin_client to os_system_admin and downstream srbac tests failures.\n\nChange-Id: I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/e32aa10c85e910b616420a89665743f2a23b78d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/e32aa10c85e910b616420a89665743f2a23b78d1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6d3dc86944a4a5c39c92bbac6baafd3668dddc9e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-24 09:13:49.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/06/884006/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/06/884006/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/4"}}},"commit":{"parents":[{"commit":"87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b","subject":"Merge \"Refactor rbac tests structure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/87df2f444b7e9e83dc56fbe7d5ac27c1c02fe37b"}]}],"author":{"name":"msava","email":"msava@redhat.com","date":"2023-05-24 08:17:31.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2023-05-24 09:13:49.000000000","tz":0},"subject":"Fix admin client to system admin","message":"Fix admin client to system admin\n\nGlance now support system scope.This patch fix admin_client to os_system_admin\n\nChange-Id: I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/6d3dc86944a4a5c39c92bbac6baafd3668dddc9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/6d3dc86944a4a5c39c92bbac6baafd3668dddc9e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cf1d8a4071ac98d77c7eb93fd30c47106be739bc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-12 20:42:07.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"real_uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/06/884006/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/06/884006/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/06/884006/5"}}},"commit":{"parents":[{"commit":"c6bf89f65815ed32560d7f2379a82e01ceee3ca8","subject":"Support py3.12 and drop py3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/c6bf89f65815ed32560d7f2379a82e01ceee3ca8"}]}],"author":{"name":"msava","email":"msava@redhat.com","date":"2023-05-24 08:17:31.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2024-11-12 20:42:07.000000000","tz":0},"subject":"Fix admin client to system admin","message":"Fix admin client to system admin\n\nGlance now support system scope.This patch fix admin_client to os_system_admin\n\nChange-Id: I45cfb8cbcef97ca326e6ace9d28815f3b641ad9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/cf1d8a4071ac98d77c7eb93fd30c47106be739bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/cf1d8a4071ac98d77c7eb93fd30c47106be739bc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
