)]}'
{"id":"openstack%2Fglance-tempest-plugin~966364","triplet_id":"openstack%2Fglance-tempest-plugin~master~I04780d98da6fc49fc852a0daeb18e212cabd7cc3","project":"openstack/glance-tempest-plugin","branch":"master","topic":"2024-1-last","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-12-08 15:32:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-12-08 16:11:46.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-12-08 16:11:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I04780d98da6fc49fc852a0daeb18e212cabd7cc3","subject":"Update stable branch jobs","status":"MERGED","created":"2025-11-06 22:15:32.000000000","updated":"2025-12-08 16:11:46.000000000","submitted":"2025-12-08 16:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966364-2024-1-last","meta_rev_id":"0dd52c4950e13a5fa1090a782b642f19bc6281a7","_number":966364,"virtual_id_number":966364,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-08 16:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2025-12-08 15:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-11-23 15:15:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2025-12-08 15:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 23:03:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-23 13:48:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-11-25 19:29:01.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-11-25 19:29:01.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"a87d19e472e585fc70b042a6829f9fa2c027e9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 22:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04994f130fa2ce7cb586dc9b9de8f0c5890f30a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 23:03:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/932768db3e9044a78bb8970f818ab8d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026762356b95428d82c6e6fb2d33be2c : SUCCESS in 2m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5da2256593314e0b8789cadaafef61b0 : SUCCESS in 33m 41s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/17c81c01c57b422dabf9e598184c204d : SUCCESS in 36m 03s\n- glance-secure-rbac-protection-functional-2025-2 https://zuul.opendev.org/t/openstack/build/37ce97551ac54806a9c328511e143bd2 : SUCCESS in 34m 47s\n- glance-secure-rbac-protection-functional-2025-1 https://zuul.opendev.org/t/openstack/build/9080df4285224418bd87301493f169fd : FAILURE in 30m 04s\n- glance-secure-rbac-protection-functional-2024-2 https://zuul.opendev.org/t/openstack/build/57bef76b0f9c4e45a3b8e7426faeaa36 : FAILURE in 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"23f5acc5fe728ce35a8a59daf544b6577a0eed1e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 03:43:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"060168061c45a69bd24af4136699fd2791361575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 04:36:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b966c73c69b4ef1b24b3fbfd2db9c54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a25440ed244a74b8ee0a32a2ed3f24 : SUCCESS in 2m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc80e29e841340a99f2e7672f0acaa78 : SUCCESS in 37m 55s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc350ad165914b42ba9c8b7018fa09c1 : SUCCESS in 26m 22s\n- glance-secure-rbac-protection-functional-2025-2 https://zuul.opendev.org/t/openstack/build/bb7345dfc03841d5ac64cb366ddcb052 : SUCCESS in 35m 40s\n- glance-secure-rbac-protection-functional-2025-1 https://zuul.opendev.org/t/openstack/build/acc061290453407e8d4185c800114884 : SUCCESS in 33m 18s\n- glance-secure-rbac-protection-functional-2024-2 https://zuul.opendev.org/t/openstack/build/7380623a3358479fa9a9fe223980f043 : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e73c255aebfdaf6e502d55dcee747803cb330817","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 13:48:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab87aaf7fc16543d8b6316de4242d1901fac6d54","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 13:48:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b0541718a60543ab519d4764b322918f3433e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 13:48:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":2},{"id":"257c64d5fc12be8c1e3e2ac1e2e767ede17dbdaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 14:30:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be72ca119f9e48468183f2a50456c765\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690bd0536ed34c699b311a3e4b70ce68 : SUCCESS in 1m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/902ce6f37be54a36b6dbd5ae8736b940 : SUCCESS in 32m 45s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c897cdd998984c28950a263960ec2a5c : SUCCESS in 22m 55s\n- glance-secure-rbac-protection-functional-2025-1 https://zuul.opendev.org/t/openstack/build/5e881e8cfe1d4606981b7e4a2b38cf8f : FAILURE in 35m 33s\n- glance-secure-rbac-protection-functional-2024-2 https://zuul.opendev.org/t/openstack/build/49cc097c5a464530bf23b7e3cd494bb8 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7ca02458b5c5a1c1f7e0bd2af790965a572acc29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 15:15:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36d74d838cab1ccfa3eae26128aa84d5612df3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 15:53:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3627f689bb4e4f539163712601dc320f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45484920ec1d4c99bb805af9a1d25599 : SUCCESS in 2m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1012b3ad4fa4e578c3b132a48f7140e : SUCCESS in 34m 25s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b440206ea3e94bc28b263e896139c615 : SUCCESS in 19m 48s\n- glance-secure-rbac-protection-functional-2025-1 https://zuul.opendev.org/t/openstack/build/28bc64a5d52b4640adc95eb373fc9f0b : SUCCESS in 33m 16s\n- glance-secure-rbac-protection-functional-2024-2 https://zuul.opendev.org/t/openstack/build/c10cc1a585564d43b812af938ad12720 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":2},{"id":"44228a56ce9a011409993fc3a57631fd6cbc6e8d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-08 15:32:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd717cc9d7bcf1188987bd58a60298e37af27917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 15:33:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18c99ca6a6ad38387854e8fbbbc8688e9b4431ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:11:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e3bbe2e6514eb38deb7446eb998042\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b943d8c72694f88ad4fe3273f348774 : SUCCESS in 1m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de933b72b355463285d7af44a19fa2f8 : SUCCESS in 34m 14s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/796bfc5b377d4e0c9331b96c3a4596f8 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0dd52c4950e13a5fa1090a782b642f19bc6281a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:11:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac4115ea269752cc6d9e418fcc46b9c87664af6d","revisions":{"e0531c3bc039d5305102b860c2722fd1568bd5a7":{"kind":"REWORK","_number":1,"created":"2025-11-06 22:15:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/966364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/64/966364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/1"}}},"commit":{"parents":[{"commit":"620d3247994662d3eeea6dd8b094f8962f0d25b2","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/620d3247994662d3eeea6dd8b094f8962f0d25b2"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:40:40.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 22:11:18.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs.\n\nChange-Id: I04780d98da6fc49fc852a0daeb18e212cabd7cc3\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/e0531c3bc039d5305102b860c2722fd1568bd5a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/e0531c3bc039d5305102b860c2722fd1568bd5a7"}]},"branch":"refs/heads/master"},"ac4115ea269752cc6d9e418fcc46b9c87664af6d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-23 13:48:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/966364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance-tempest-plugin","ref":"refs/changes/64/966364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance-tempest-plugin refs/changes/64/966364/2"}}},"commit":{"parents":[{"commit":"be482976001ba5e946029ddd0e3fe8ee1852568d","subject":"Merge \"Add srbac metadef tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/be482976001ba5e946029ddd0e3fe8ee1852568d"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 21:40:40.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-11-23 13:48:55.000000000","tz":0},"subject":"Update stable branch jobs","message":"Update stable branch jobs\n\nRemove unmaintained/2024.1 jobs.\n\nChange-Id: I04780d98da6fc49fc852a0daeb18e212cabd7cc3\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/ac4115ea269752cc6d9e418fcc46b9c87664af6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance-tempest-plugin/commit/ac4115ea269752cc6d9e418fcc46b9c87664af6d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
