)]}'
{"id":"openstack%2Fglance_store~1002152","triplet_id":"openstack%2Fglance_store~master~If16529fc2d268a37932a94c33a191e63ccad1cae","project":"openstack/glance_store","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-26 19:15:33.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If16529fc2d268a37932a94c33a191e63ccad1cae","subject":"Remove unused with_venv script","status":"NEW","created":"2026-08-24 15:49:55.000000000","updated":"2026-08-30 07:04:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4bce4c333afb54e1c73a8c95806e3003990c4da","_number":1002152,"virtual_id_number":1002152,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 07:04:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 18:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46a7829a9aa96f9be14d1417a9014b32e657437e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 15:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0683eb139433563287313ffddfffd86cf85ad23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:45:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c5f4199933e4ccdab03064592e254e5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5739778218e4949bd60f9b3ab7187df : SUCCESS in 1h 47m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993bfbc5bdbd4bf997830902171b62ee : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfe433ab03b44fe78f1c8e59cf232f19 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66fb7c1477aa4d5db47467ade2146ef7 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07474317e5f84a7a8453c0ce663ce931 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4800fa17f6a40fd8adb9af306180be2 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e7cbece5d574868b588f103ac653baf : SUCCESS in 3m 51s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/e0b9fcf6460841199dde3333e1fa869d : SUCCESS in 10m 14s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/47ac0ab54791442e8161e24dc59e753d : FAILURE in 2h 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2acf4a6e20c9dd2510d10e3168f3cd4dcd3f63cf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 16:35:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be44d1e2c67f1c0f25db7c778fe9b75a38b4f578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:15:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4828d85485b2434289bb39cca6c4bf44\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/657721233fde4aa399f3c1400164a107 : SUCCESS in 1h 39m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae05174317374dc4a2ae5af41f99ce87 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/926c67b9f1514d55b5f77c1accb1aa88 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb0f9a8d4cfb438e9f7d8fd03d09bca0 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7729f96092c4423f9aeba81fd2b99777 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ba0ffb7fdf4d2fbd9688febda14014 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/737de5e886ec44dcaf18c1d16025c103 : SUCCESS in 2m 11s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/1db51f913f8a4d4490b535661cdcc033 : SUCCESS in 5m 09s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/647df893687c4ee2b16961c1514b0a3f : FAILURE in 2h 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"528efbe618e767e41053da1bfa901716148cc6e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-29 16:57:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"92ed1ab0b9d8737d18d9e059ac7a2bc741c9f590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 18:51:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f9fe7155e0b44719288be4d737a2cfc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f822676d14942a9b22fecb5355613a0 : SUCCESS in 1h 30m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd13ecfb3f1c4968b76bc3ddcf478946 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23f0c4a24a45408e8ec3c47a86ccd1bd : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78f2b1de9521499d9ab5d2e117cdbc8a : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7608cfd48514b1999685e834e08309c : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e00767191b04889ac6a479f6b2b5dca : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26ee7904216b44b799c14b43b1f982fc : SUCCESS in 4m 28s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/5c17f068d710468bacddc50083d20925 : SUCCESS in 8m 52s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b7dac4d43ef43e7b16c4512f4d77204 : SUCCESS in 1h 46m 15s","accounts_in_message":[],"_revision_number":2},{"id":"62544c2b2b310bcf36731ce8290f9708532f88a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-29 19:25:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"03f6d53f54f23f7336e9d061f4bf027ae91dd147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 21:23:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5486d88b59e48f3af5a36f55e8ebaa9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89af3df3da4f434889437e1841979cda : SUCCESS in 45m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1984118894544ce9ad212e612c47d34d : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f781e2be30d4786a9ad91a414aa7f40 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb15dff6a43c4ba28a4ece4f56781470 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ffdab61f7024453a3416dcd798c756f : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f90a428396455aab7016ace0f30877 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f63cad26ab847ffb9fa71cbd817af6b : SUCCESS in 3m 29s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/e772af0e5cf942de9a38434719e82760 : SUCCESS in 5m 36s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c13be911ab6a45f1afe042aed799368b : FAILURE in 1h 55m 14s","accounts_in_message":[],"_revision_number":3},{"id":"f9e4cf29bb8e08f710fae9de73561cf6b167b36e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 04:31:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d0acd9eadea25fbf4975cac3a9df4c03f8c02b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-30 05:42:00.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f4bce4c333afb54e1c73a8c95806e3003990c4da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 07:04:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e4ec69a468f45498082a95fa30b315d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c293fc541dc4b73a4ebd23276756585 : SUCCESS in 1h 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592cb1ac288d4f18b71046f4878d79df : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b12dfcf404524bd893cd9296eb3ab110 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bb5d66acb9c455fb68f28a53c45ea46 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8bd8e16a33c442ffbf2fabb9e6c24e12 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d684b971ce4ad0a0489b1e7890d092 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ffd3cfef7834deaacbbcd1aa066470c : SUCCESS in 3m 52s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/5bf123bad757481fbdfe6aedbd1ab18f : SUCCESS in 7m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d7c3edf74944e7aa544f98d3898ba24 : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"89cd33c17b4e19fd96f4e4fe30316eaad8db0011","revisions":{"b816ba11743193e8cd1714f6d8158f24287e8246":{"kind":"REWORK","_number":1,"created":"2026-08-24 15:49:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1002152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/52/1002152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/1"}}},"commit":{"parents":[{"commit":"8ff4b18d85b990d1a0a7d66bdf62421dd00a6118","subject":"Skip functional tests for git config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/8ff4b18d85b990d1a0a7d66bdf62421dd00a6118"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 15:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 15:49:26.000000000","tz":540},"subject":"Remove unused with_venv script","message":"Remove unused with_venv script\n\nIt was used by the legacy test execution script, which was already\nremoved[1].\n\n[1] 14a188fa90fc31fe9bb756d1071ba4df0ab0d0a6\n\nChange-Id: If16529fc2d268a37932a94c33a191e63ccad1cae\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b816ba11743193e8cd1714f6d8158f24287e8246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b816ba11743193e8cd1714f6d8158f24287e8246"}]},"branch":"refs/heads/master"},"059885bd579208a88f186ad2fbc2c2369668afb1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-29 16:57:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1002152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/52/1002152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/2"}}},"commit":{"parents":[{"commit":"d46dfc1b7a63a52fa9ffd48e8a52aa7ede54f3ad","subject":"Skip functional tests for git config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d46dfc1b7a63a52fa9ffd48e8a52aa7ede54f3ad"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 15:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-29 16:57:01.000000000","tz":0},"subject":"Remove unused with_venv script","message":"Remove unused with_venv script\n\nIt was used by the legacy test execution script, which was already\nremoved[1].\n\n[1] 14a188fa90fc31fe9bb756d1071ba4df0ab0d0a6\n\nChange-Id: If16529fc2d268a37932a94c33a191e63ccad1cae\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/059885bd579208a88f186ad2fbc2c2369668afb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/059885bd579208a88f186ad2fbc2c2369668afb1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8ff4b18d85b990d1a0a7d66bdf62421dd00a6118","ours":"b816ba11743193e8cd1714f6d8158f24287e8246","theirs":"d46dfc1b7a63a52fa9ffd48e8a52aa7ede54f3ad","merge_strategy":"recursive","contains_conflicts":false}},"2e9c3845a09dbc574a81fd9315ddd77ffe38d157":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-29 19:25:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1002152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/52/1002152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/3"}}},"commit":{"parents":[{"commit":"bbf88d15c65d464d5a30db700ae9b34e9b410a92","subject":"Skip functional tests for git config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/bbf88d15c65d464d5a30db700ae9b34e9b410a92"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 15:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-29 19:25:47.000000000","tz":0},"subject":"Remove unused with_venv script","message":"Remove unused with_venv script\n\nIt was used by the legacy test execution script, which was already\nremoved[1].\n\n[1] 14a188fa90fc31fe9bb756d1071ba4df0ab0d0a6\n\nChange-Id: If16529fc2d268a37932a94c33a191e63ccad1cae\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2e9c3845a09dbc574a81fd9315ddd77ffe38d157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2e9c3845a09dbc574a81fd9315ddd77ffe38d157"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d46dfc1b7a63a52fa9ffd48e8a52aa7ede54f3ad","ours":"059885bd579208a88f186ad2fbc2c2369668afb1","theirs":"bbf88d15c65d464d5a30db700ae9b34e9b410a92","merge_strategy":"recursive","contains_conflicts":false}},"6ed36b6518ca7ffb07cea637bacfc7b1604e8ef6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-30 04:31:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1002152/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/52/1002152/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/4"}}},"commit":{"parents":[{"commit":"9e77597685f03c6b950b31af557f508718b471e7","subject":"Skip functional tests for git config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9e77597685f03c6b950b31af557f508718b471e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 15:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 04:31:36.000000000","tz":0},"subject":"Remove unused with_venv script","message":"Remove unused with_venv script\n\nIt was used by the legacy test execution script, which was already\nremoved[1].\n\n[1] 14a188fa90fc31fe9bb756d1071ba4df0ab0d0a6\n\nChange-Id: If16529fc2d268a37932a94c33a191e63ccad1cae\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6ed36b6518ca7ffb07cea637bacfc7b1604e8ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6ed36b6518ca7ffb07cea637bacfc7b1604e8ef6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"bbf88d15c65d464d5a30db700ae9b34e9b410a92","ours":"2e9c3845a09dbc574a81fd9315ddd77ffe38d157","theirs":"9e77597685f03c6b950b31af557f508718b471e7","merge_strategy":"recursive","contains_conflicts":false}},"89cd33c17b4e19fd96f4e4fe30316eaad8db0011":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-30 05:42:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1002152/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/52/1002152/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/52/1002152/5"}}},"commit":{"parents":[{"commit":"2f12b73f2fe48a3c6e0b304d57c13f9bf348e99e","subject":"Skip functional tests for git config files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2f12b73f2fe48a3c6e0b304d57c13f9bf348e99e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 15:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-30 05:42:00.000000000","tz":0},"subject":"Remove unused with_venv script","message":"Remove unused with_venv script\n\nIt was used by the legacy test execution script, which was already\nremoved[1].\n\n[1] 14a188fa90fc31fe9bb756d1071ba4df0ab0d0a6\n\nChange-Id: If16529fc2d268a37932a94c33a191e63ccad1cae\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/89cd33c17b4e19fd96f4e4fe30316eaad8db0011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/89cd33c17b4e19fd96f4e4fe30316eaad8db0011"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9e77597685f03c6b950b31af557f508718b471e7","ours":"6ed36b6518ca7ffb07cea637bacfc7b1604e8ef6","theirs":"2f12b73f2fe48a3c6e0b304d57c13f9bf348e99e","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
