)]}'
{"id":"openstack%2Fglance_store~578549","triplet_id":"openstack%2Fglance_store~master~I875f0c45a3b47aa6ee1fbdba89d807a649ce6321","project":"openstack/glance_store","branch":"master","topic":"glance_retries","attention_set":{"1297":{"account":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"last_update":"2022-08-09 16:40:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"11827":{"account":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"last_update":"2022-08-09 16:40:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I875f0c45a3b47aa6ee1fbdba89d807a649ce6321","subject":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts","status":"NEW","created":"2018-06-27 20:07:31.000000000","updated":"2022-08-09 16:40:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"554c3cab49760806633b51d6541758f57d90a38a","_number":578549,"virtual_id_number":578549,"owner":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":1,"date":"2018-12-12 15:51:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-02-15 15:06:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":-2,"date":"2019-02-22 23:34:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},{"value":1,"date":"2019-02-22 05:45:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":-2,"date":"2022-08-09 16:40:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},{"value":1,"date":"2019-02-22 05:45:48.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-27 20:25:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2018-12-10 19:33:38.000000000","updated_by":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"reviewer":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"state":"REVIEWER"},{"updated":"2018-12-11 19:58:08.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-12-12 15:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-15 15:06:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-02-22 05:45:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-02-22 23:34:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-09 16:40:18.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"1dad7b5c4bc2b04addced991e66a3f085ac6facd","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2018-06-27 20:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83425747f43791985d542ba974de4b77ceb6b0a7","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2018-06-27 20:09:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e2cb1a280efc704960e5f26530abab2c77bcf96","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2018-06-27 20:25:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b4b64540175bdfe55838d2c8e067f407cedacd6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-27 22:29:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/49/578549/2/check/openstack-tox-pep8/1773d78/ : FAILURE in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/49/578549/2/check/openstack-tox-py27/ea21e1f/ : FAILURE in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/49/578549/2/check/openstack-tox-py35/ac2a44b/ : FAILURE in 5m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/49/578549/2/check/build-openstack-sphinx-docs/4120626/html/ : SUCCESS in 6m 24s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/49/578549/2/check/legacy-tempest-dsvm-neutron-src/c351f92/ : SUCCESS in 1h 34m 38s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/49/578549/2/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/de9ea52/ : SUCCESS in 1h 12m 31s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/49/578549/2/check/legacy-tempest-dsvm-full-sheepdog/bbc6960/ : FAILURE in 1h 36m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/49/578549/2/check/openstack-tox-lower-constraints/a2a984a/ : FAILURE in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6fe5c3474203223ea86930ddeaa0ce0dd7dd01c3","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2018-07-12 00:21:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a63c46607145125c7588c310710f43b89692ca72","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2018-07-12 00:23:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b224775b11cd8b81efc47a7da3e59d4dc9a8fce","author":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"date":"2018-12-10 19:33:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"78d7f0e9f6c3796e6ff452ad6e5c39184ff2283a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-10 22:06:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/49/578549/3/check/openstack-tox-pep8/1d78178/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/49/578549/3/check/openstack-tox-py27/8cb752e/ : FAILURE in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/49/578549/3/check/openstack-tox-py35/75fc643/ : FAILURE in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/49/578549/3/check/openstack-tox-py36/873f4e9/ : FAILURE in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/49/578549/3/check/openstack-tox-docs/5b7b2b8/html/ : SUCCESS in 3m 54s\n- tempest-full http://logs.openstack.org/49/578549/3/check/tempest-full/6b92bbf/ : SUCCESS in 1h 23m 11s\n- tempest-full-py3 http://logs.openstack.org/49/578549/3/check/tempest-full-py3/649f211/ : SUCCESS in 1h 10m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/578549/3/check/openstack-tox-lower-constraints/092e76f/ : FAILURE in 4m 36s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/49/578549/3/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/1d232d9/ : SUCCESS in 2h 24m 19s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/49/578549/3/check/legacy-tempest-dsvm-full-sheepdog/951df32/ : FAILURE in 1h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c4b4c16754bb0a9b69d7b9f5e7fa022b5427b7b","author":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"date":"2018-12-11 18:10:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"971f4552ef20275dec06a01a6d1c5f4af1d86a69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 19:57:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/49/578549/4/check/openstack-tox-pep8/7de6859/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/49/578549/4/check/openstack-tox-py27/1e36ce8/ : FAILURE in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/49/578549/4/check/openstack-tox-py35/78a9c63/ : FAILURE in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/49/578549/4/check/openstack-tox-py36/a66fda5/ : FAILURE in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/49/578549/4/check/openstack-tox-docs/bf7d9d4/html/ : SUCCESS in 4m 02s\n- tempest-full http://logs.openstack.org/49/578549/4/check/tempest-full/5648004/ : SUCCESS in 1h 25m 47s\n- tempest-full-py3 http://logs.openstack.org/49/578549/4/check/tempest-full-py3/78cdd2a/ : FAILURE in 1h 04m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/578549/4/check/openstack-tox-lower-constraints/4239803/ : FAILURE in 4m 34s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/49/578549/4/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/a4334b6/ : SUCCESS in 1h 26m 08s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/49/578549/4/check/legacy-tempest-dsvm-full-sheepdog/3167665/ : SUCCESS in 1h 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a00febb2f62353f6a01049edd8db29498fd0528a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-11 19:58:08.000000000","message":"Patch Set 4:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- openstack-tox-lower-constraints: unrecognized error\n- openstack-tox-py27: unrecognized error\n- openstack-tox-py35: unrecognized error\n- openstack-tox-py36: unrecognized error\n- tempest-full-py3: https://bugs.launchpad.net/bugs/1737634\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"993c3effbb4f91fdf1828d450a9db9275181b133","author":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"date":"2018-12-12 14:20:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d954db7c41c77795e6a85606af3227940dc59d89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 15:51:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/49/578549/5/check/openstack-tox-pep8/03f63c6/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/49/578549/5/check/openstack-tox-py27/2c6a7c2/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/49/578549/5/check/openstack-tox-py35/6560b43/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/49/578549/5/check/openstack-tox-py36/3d892ce/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/49/578549/5/check/openstack-tox-docs/4431202/html/ : SUCCESS in 4m 22s\n- tempest-full http://logs.openstack.org/49/578549/5/check/tempest-full/cc408a4/ : SUCCESS in 1h 16m 58s\n- tempest-full-py3 http://logs.openstack.org/49/578549/5/check/tempest-full-py3/77d1de4/ : SUCCESS in 1h 12m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/578549/5/check/openstack-tox-lower-constraints/f510281/ : SUCCESS in 4m 14s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/49/578549/5/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/33a5ce7/ : SUCCESS in 1h 23m 29s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/49/578549/5/check/legacy-tempest-dsvm-full-sheepdog/2dd1470/ : SUCCESS in 1h 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"50e8618c8eb7181b2eebecc5558ed36079cddf29","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-02-15 15:06:12.000000000","message":"Patch Set 5: Code-Review+2\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":5},{"id":"b0841663c7308bcc6a27515284e326f68862b066","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-02-22 05:45:48.000000000","message":"Patch Set 5: Code-Review+1 Review-Priority+1\n\nCode looks good, can we have a unit test to cover this change?","accounts_in_message":[],"_revision_number":5},{"id":"dcbf4329d22f9aa7f877d20a37358c0ce93930f1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-02-22 23:34:07.000000000","message":"Patch Set 5: Code-Review-2\n\nDoesn\u0027t the BufferedReader do this already?\nhttp://git.openstack.org/cgit/openstack/glance_store/tree/glance_store/_drivers/swift/buffered.py\n\nIf not, you can use the TestBufferedReader class in test_swift_store to get some testing ideas.","accounts_in_message":[],"_revision_number":5},{"id":"554c3cab49760806633b51d6541758f57d90a38a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-08-09 16:40:18.000000000","message":"Patch Set 5: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3dc4db82fff4285a821add77055da4c1c342740a","revisions":{"b97e1e7f5316eb7f52ea584e623d845d310ef2ce":{"kind":"REWORK","_number":1,"created":"2018-06-27 20:07:31.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/49/578549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/578549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/578549/1"}}},"commit":{"parents":[{"commit":"5a20d47cd3a6e16feea602a282b2b7892eb5b7d5","subject":"Deprecate store_capabilities_update_min_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5a20d47cd3a6e16feea602a282b2b7892eb5b7d5"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:06:35.000000000","tz":-420},"committer":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:07:30.000000000","tz":-420},"subject":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts","message":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts\n\nThis is useful when the network to swift is flakey, or swift\nis flakey, or any of the above. It also matches the retrying\nof gets so it adds a nice symmetry.\n\nChange-Id: I875f0c45a3b47aa6ee1fbdba89d807a649ce6321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b97e1e7f5316eb7f52ea584e623d845d310ef2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b97e1e7f5316eb7f52ea584e623d845d310ef2ce"}]},"branch":"refs/heads/master"},"d263c09972940b5180fcce807a86bb0b6d3463e0":{"kind":"REWORK","_number":2,"created":"2018-06-27 20:09:25.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/49/578549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/578549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/578549/2"}}},"commit":{"parents":[{"commit":"5a20d47cd3a6e16feea602a282b2b7892eb5b7d5","subject":"Deprecate store_capabilities_update_min_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5a20d47cd3a6e16feea602a282b2b7892eb5b7d5"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:06:35.000000000","tz":-420},"committer":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:09:16.000000000","tz":-420},"subject":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts","message":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts\n\nThis is useful when the network to swift is flakey, or swift\nis flakey, or any of the above. It also matches the retrying\nof gets so it adds a nice symmetry.\n\nChange-Id: I875f0c45a3b47aa6ee1fbdba89d807a649ce6321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d263c09972940b5180fcce807a86bb0b6d3463e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d263c09972940b5180fcce807a86bb0b6d3463e0"}]},"branch":"refs/heads/master"},"17b6c8e78075f27046b4594c59f78cdd5ff04760":{"kind":"REWORK","_number":3,"created":"2018-12-10 19:33:38.000000000","uploader":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"ref":"refs/changes/49/578549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/578549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/578549/3"}}},"commit":{"parents":[{"commit":"b8b873433e4dba9fae9cd8c51a0a7b0079875d0e","subject":"Merge \"Provision to add new config options in sample config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b8b873433e4dba9fae9cd8c51a0a7b0079875d0e"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:06:35.000000000","tz":-420},"committer":{"name":"sorrowless","email":"sbog@sbog.ru","date":"2018-12-10 19:33:27.000000000","tz":180},"subject":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts","message":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts\n\nThis is useful when the network to swift is flakey, or swift\nis flakey, or any of the above. It also matches the retrying\nof gets so it adds a nice symmetry.\n\nChange-Id: I875f0c45a3b47aa6ee1fbdba89d807a649ce6321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/17b6c8e78075f27046b4594c59f78cdd5ff04760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/17b6c8e78075f27046b4594c59f78cdd5ff04760"}]},"branch":"refs/heads/master"},"f6e637002a76acd913e6c03fa039674c71636c08":{"kind":"REWORK","_number":4,"created":"2018-12-11 18:10:41.000000000","uploader":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"ref":"refs/changes/49/578549/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/578549/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/578549/4"}}},"commit":{"parents":[{"commit":"b8b873433e4dba9fae9cd8c51a0a7b0079875d0e","subject":"Merge \"Provision to add new config options in sample config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b8b873433e4dba9fae9cd8c51a0a7b0079875d0e"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:06:35.000000000","tz":-420},"committer":{"name":"sorrowless","email":"sbog@sbog.ru","date":"2018-12-11 18:09:56.000000000","tz":180},"subject":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts","message":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts\n\nThis is useful when the network to swift is flakey, or swift\nis flakey, or any of the above. It also matches the retrying\nof gets so it adds a nice symmetry.\n\nChange-Id: I875f0c45a3b47aa6ee1fbdba89d807a649ce6321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f6e637002a76acd913e6c03fa039674c71636c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f6e637002a76acd913e6c03fa039674c71636c08"}]},"branch":"refs/heads/master"},"3dc4db82fff4285a821add77055da4c1c342740a":{"kind":"REWORK","_number":5,"created":"2018-12-12 14:20:08.000000000","uploader":{"_account_id":11827,"name":"Stanislaw Bogatkin","email":"regs+ubuntu.com@sbog.ru","username":"sbogatkin"},"ref":"refs/changes/49/578549/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/578549/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/578549/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/578549/5"}}},"commit":{"parents":[{"commit":"b8b873433e4dba9fae9cd8c51a0a7b0079875d0e","subject":"Merge \"Provision to add new config options in sample config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b8b873433e4dba9fae9cd8c51a0a7b0079875d0e"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2018-06-27 20:06:35.000000000","tz":-420},"committer":{"name":"sorrowless","email":"sbog@sbog.ru","date":"2018-12-12 14:19:54.000000000","tz":180},"subject":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts","message":"Add \u0027swift_store_retry_put_count\u0027 for retrying chunk puts\n\nThis is useful when the network to swift is flakey, or swift\nis flakey, or any of the above. It also matches the retrying\nof gets so it adds a nice symmetry.\n\nChange-Id: I875f0c45a3b47aa6ee1fbdba89d807a649ce6321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3dc4db82fff4285a821add77055da4c1c342740a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3dc4db82fff4285a821add77055da4c1c342740a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"failing_atoms":[],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
