)]}'
{"id":"openstack%2Fglance_store~651941","triplet_id":"openstack%2Fglance_store~master~Iee3b49235be2af64eac3933e41c7e44fc1a564f4","project":"openstack/glance_store","branch":"master","topic":"bug/1818915","hashtags":[],"change_id":"Iee3b49235be2af64eac3933e41c7e44fc1a564f4","subject":"Python3: Fix return type on CooperativeReader.read","status":"MERGED","created":"2019-04-11 23:41:42.000000000","updated":"2019-05-15 13:16:08.000000000","submitted":"2019-05-15 13:16:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651941-1557926168533-0de1d12d","meta_rev_id":"8251e65d98528fd8478860c6298d29c780735144","_number":651941,"virtual_id_number":651941,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":2,"date":"2019-05-15 13:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-04-17 16:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-05-15 11:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-05-15 11:13:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 16:15:55.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-05-15 11:13:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-05-15 13:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e452860fb4e39a664a8882f35a869ed29d954f8e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-04-11 23:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecfe676fb790d94d7836ad4a555b1f4cd3595282","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 01:44:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/41/651941/1/check/tempest-full/353ad2c/ : SUCCESS in 2h 00m 00s\n- tempest-full-py3 http://logs.openstack.org/41/651941/1/check/tempest-full-py3/aad2d3f/ : SUCCESS in 1h 39m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/651941/1/check/openstack-tox-lower-constraints/2e3f865/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/41/651941/1/check/openstack-tox-pep8/31d0a44/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/41/651941/1/check/openstack-tox-py27/7a7e65c/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/41/651941/1/check/openstack-tox-py35/d3e3b5d/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/41/651941/1/check/openstack-tox-py36/509752e/ : SUCCESS in 8m 47s\n- openstack-tox-py37 http://logs.openstack.org/41/651941/1/check/openstack-tox-py37/20338f3/ : SUCCESS in 8m 04s\n- openstack-tox-docs http://logs.openstack.org/41/651941/1/check/openstack-tox-docs/0cad737/html/ : SUCCESS in 4m 39s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/41/651941/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/7c07e24/ : SUCCESS in 1h 30m 04s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/41/651941/1/check/legacy-tempest-dsvm-full-sheepdog/36d124e/ : FAILURE in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"431bc000224991c60b83e28c30afe0243eb525ac","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-04-12 12:34:38.000000000","message":"Patch Set 1:\n\nThanks Cyril for taking this on. Only thing I\u0027d like to see is a test for this to make sure we do not slip back. Should be fairly easy unit test to make, just request zero read from the function and make sure that the return is indeed empty byte","accounts_in_message":[],"_revision_number":1},{"id":"7b14a334c69ecc26a7c3e8b85aed64439e962e0f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-04-12 12:36:05.000000000","message":"Patch Set 1:\n\nSorry, istead of empty read it should be even easier to just mock the reader and have it raising the StopIteration.","accounts_in_message":[],"_revision_number":1},{"id":"743506d4d694300cf024803fca532c83eba4d25b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-04-12 12:47:05.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"0a8df10d29ee7206f55ebc52fbd90ff97f3d8ffb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-04-12 21:17:49.000000000","message":"Patch Set 1:\n\nJust reading from a temporary file seems to lead us to the faulty code, so it is even simpler :)","accounts_in_message":[],"_revision_number":1},{"id":"b5401e16c592787285b67e2e6928bff88ea67d28","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2019-04-12 21:18:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f72e2dff15b3ebfa5f270aa7f97bac62d5ad6216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 23:27:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/41/651941/2/check/tempest-full/a267b6a/ : SUCCESS in 1h 56m 08s\n- tempest-full-py3 http://logs.openstack.org/41/651941/2/check/tempest-full-py3/3f6b5c1/ : SUCCESS in 1h 33m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/651941/2/check/openstack-tox-lower-constraints/1a33ae1/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/41/651941/2/check/openstack-tox-pep8/dfa8992/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/41/651941/2/check/openstack-tox-py27/5d3c63f/ : SUCCESS in 5m 41s\n- openstack-tox-py35 http://logs.openstack.org/41/651941/2/check/openstack-tox-py35/416cd0a/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/41/651941/2/check/openstack-tox-py36/d736a33/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/41/651941/2/check/openstack-tox-py37/466a2a5/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/41/651941/2/check/openstack-tox-docs/fab09c9/html/ : SUCCESS in 6m 29s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/41/651941/2/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/5058843/ : SUCCESS in 1h 19m 16s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/41/651941/2/check/legacy-tempest-dsvm-full-sheepdog/f91a039/ : FAILURE in 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00c20fd14cceea7fef3a99d94af20f9309983e2a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-04-17 16:15:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"89107472771e5a2e6241c4f806c7d042cf4fef9e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-15 11:13:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11379a93096fde7e1065f6ebf0c16822264c776f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 11:14:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b03db724f37645ad2a50cda03a5b46f20c9cd469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:16:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"83071024e832da66f30c00051dfff2051cf9d308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:16:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/41/651941/2/gate/tempest-full/5557ce6/ : SUCCESS in 1h 58m 52s\n- tempest-full-py3 http://logs.openstack.org/41/651941/2/gate/tempest-full-py3/31c5f23/ : SUCCESS in 1h 27m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/651941/2/gate/openstack-tox-lower-constraints/1e96051/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/41/651941/2/gate/openstack-tox-pep8/980fb16/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/41/651941/2/gate/openstack-tox-py27/cc45d7f/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/41/651941/2/gate/openstack-tox-py35/2ae65ca/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/41/651941/2/gate/openstack-tox-py36/716d098/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/41/651941/2/gate/openstack-tox-py37/959852a/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/41/651941/2/gate/openstack-tox-docs/e43acdd/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a3166558f1b2f382075145502576919cecec61e","revisions":{"a3dcb80d25ae70332e877d1e992c26e4baf6aac8":{"kind":"REWORK","_number":1,"created":"2019-04-11 23:41:42.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/41/651941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/41/651941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/41/651941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/41/651941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/41/651941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/41/651941/1"}}},"commit":{"parents":[{"commit":"ba6b5cf2ae0afb69bf31ab1e2b30edaeba388946","subject":"Merge \"Uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ba6b5cf2ae0afb69bf31ab1e2b30edaeba388946"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-04-11 23:40:04.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-04-11 23:40:04.000000000","tz":120},"subject":"Python3: Fix return type on CooperativeReader.read","message":"Python3: Fix return type on CooperativeReader.read\n\nCooperativeReader.read() should return bytes. A similar change was\nmerged in Glance.\n\nChange-Id: Iee3b49235be2af64eac3933e41c7e44fc1a564f4\nCloses-Bug: #1818915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a3dcb80d25ae70332e877d1e992c26e4baf6aac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a3dcb80d25ae70332e877d1e992c26e4baf6aac8"}]},"branch":"refs/heads/master"},"0a3166558f1b2f382075145502576919cecec61e":{"kind":"REWORK","_number":2,"created":"2019-04-12 21:18:07.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/41/651941/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/41/651941/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/41/651941/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/41/651941/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/41/651941/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/41/651941/2"}}},"commit":{"parents":[{"commit":"ba6b5cf2ae0afb69bf31ab1e2b30edaeba388946","subject":"Merge \"Uncap jsonschema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ba6b5cf2ae0afb69bf31ab1e2b30edaeba388946"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-04-11 23:40:04.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2019-04-12 21:17:12.000000000","tz":120},"subject":"Python3: Fix return type on CooperativeReader.read","message":"Python3: Fix return type on CooperativeReader.read\n\nCooperativeReader.read() should return bytes. A similar change was\nmerged in Glance.\n\nChange-Id: Iee3b49235be2af64eac3933e41c7e44fc1a564f4\nCloses-Bug: #1818915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0a3166558f1b2f382075145502576919cecec61e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/0a3166558f1b2f382075145502576919cecec61e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
