)]}'
{"id":"openstack%2Fglance_store~655006","triplet_id":"openstack%2Fglance_store~master~Ia0b66aac906b7920f246c4710ed06d63e11276d8","project":"openstack/glance_store","branch":"master","hashtags":[],"change_id":"Ia0b66aac906b7920f246c4710ed06d63e11276d8","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:38:04.000000000","updated":"2019-05-30 20:45:59.000000000","submitted":"2019-05-30 20:45:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655006-1559249159923-660a502c","meta_rev_id":"237d32cadf1de9ade2509fdde361d1b5309bd840","_number":655006,"virtual_id_number":655006,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-05-30 20:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2019-05-30 17:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-05-05 06:52:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-05-05 06:48:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-05-23 07:24:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-05-30 04:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2019-05-14 06:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2019-05-30 17:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 03:38:11.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-05-05 06:48:06.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-05-05 06:52:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-05-14 06:30:09.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-05-23 07:24:18.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-05-30 04:33:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-05-30 17:28:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-30 20:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06c021f3d25ab063a287b4a7a44697f9705f5bd1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:38:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e96ec36167c5093dfab7682670de2ad29f124614","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:55:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/06/655006/1/check/test-release-openstack/6b0dffe/ : SUCCESS in 4m 21s\n- tempest-full finger://ze05.openstack.org/28d3e5bf2b2a466cb732e650e5c57c2d : RETRY_LIMIT in 13m 18s\n- tempest-full-py3 http://logs.openstack.org/06/655006/1/check/tempest-full-py3/5548c26/ : RETRY_LIMIT in 19m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/655006/1/check/openstack-tox-lower-constraints/f9cf654/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/06/655006/1/check/openstack-tox-pep8/00c343d/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/06/655006/1/check/openstack-tox-py27/dcabf2f/ : SUCCESS in 5m 31s\n- openstack-tox-py35 http://logs.openstack.org/06/655006/1/check/openstack-tox-py35/34e58ad/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/06/655006/1/check/openstack-tox-py36/f004834/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/06/655006/1/check/openstack-tox-py37/193b18a/ : SUCCESS in 6m 53s\n- openstack-tox-docs http://logs.openstack.org/06/655006/1/check/openstack-tox-docs/4ba32d2/html/ : SUCCESS in 4m 46s\n- build-openstack-releasenotes http://logs.openstack.org/06/655006/1/check/build-openstack-releasenotes/3973884/html/ : SUCCESS in 4m 52s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/06/655006/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/7736356/ : RETRY_LIMIT in 17m 19s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/06/655006/1/check/legacy-tempest-dsvm-full-sheepdog/6658531/ : FAILURE in 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e453343b497eff2fcc55048a8683ce978d75d66","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 14:02:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7493826dd31a6f7d57c3482c5a9682974cd43984","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:31:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/06/655006/2/check/test-release-openstack/5ff2470/ : SUCCESS in 3m 00s\n- tempest-full finger://ze12.openstack.org/0aea1bfe85b9445880003a057e6f9c1a : RETRY_LIMIT in 23m 10s\n- tempest-full-py3 finger://ze11.openstack.org/81b14398588e48a683a35aeadc864dd7 : RETRY_LIMIT in 26m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/655006/2/check/openstack-tox-lower-constraints/b752043/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/06/655006/2/check/openstack-tox-pep8/089f031/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/06/655006/2/check/openstack-tox-py27/095bacc/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/06/655006/2/check/openstack-tox-py35/6772cca/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/06/655006/2/check/openstack-tox-py36/96fc425/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/06/655006/2/check/openstack-tox-py37/4438643/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/06/655006/2/check/openstack-tox-docs/77f9fb7/html/ : SUCCESS in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/06/655006/2/check/build-openstack-releasenotes/d7b0c1c/html/ : SUCCESS in 4m 19s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store finger://ze02.openstack.org/67118875cfd14639b1a5accd9e370b81 : RETRY_LIMIT in 21m 54s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/06/655006/2/check/legacy-tempest-dsvm-full-sheepdog/6fdc9fd/ : FAILURE in 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01d741b9b8404b6b3022fcd7934e1c9cbadcf275","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 03:38:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25aa5af806a08ec11db3fe613d3ec8311358edd0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-05 06:39:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3eef82ed709d15676d28936a2452ae26def72ad4","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-05 06:45:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7bea7d0721837fee5fa95a908569d879f0df4bd7","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-05 06:47:47.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b888a24e52d28873978942f05958f318d57bc862","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-05 06:48:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f5512c09765cfddedc6e1c1c346e4657144442f2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-05 06:52:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"73f4b575de9b607eebb191bf858da3c5afda2a84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 08:41:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/06/655006/3/check/test-release-openstack/58c81b4/ : SUCCESS in 3m 57s\n- tempest-full http://logs.openstack.org/06/655006/3/check/tempest-full/6535322/ : SUCCESS in 1h 52m 36s\n- tempest-full-py3 http://logs.openstack.org/06/655006/3/check/tempest-full-py3/a383aeb/ : SUCCESS in 1h 46m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/655006/3/check/openstack-tox-lower-constraints/0f4e416/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/06/655006/3/check/openstack-tox-pep8/ff181ca/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/06/655006/3/check/openstack-tox-py27/e78ffe9/ : SUCCESS in 5m 09s\n- openstack-tox-py35 http://logs.openstack.org/06/655006/3/check/openstack-tox-py35/094bba9/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/06/655006/3/check/openstack-tox-py36/326c50f/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/06/655006/3/check/openstack-tox-py37/95f98a8/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/06/655006/3/check/openstack-tox-docs/81e0504/html/ : SUCCESS in 3m 57s\n- build-openstack-releasenotes http://logs.openstack.org/06/655006/3/check/build-openstack-releasenotes/0cd5041/html/ : SUCCESS in 3m 45s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/06/655006/3/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/fe8d2e2/ : SUCCESS in 1h 31m 37s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/06/655006/3/check/legacy-tempest-dsvm-full-sheepdog/e86da53/ : FAILURE in 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d571560da168ba163a64cd9df72aa783f08376d0","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-05-14 06:30:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"34ff2bf6c41202b982bcff0568293d1d998b656a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-05-23 07:24:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ffe0b27f3bc2b36b148fb3456541661b9d1fd4d6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-05-30 04:33:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2fa9da75c0e481073ab13e3f0aa0e1b55d310afd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-30 17:28:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM.  May require a follow up (see comment inline).","accounts_in_message":[],"_revision_number":3},{"id":"1075913644aa3c6a34fff95e2435b037f710b7e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 17:28:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"234915fa3c57cc363fa101bcc7bc249ab0e80487","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:45:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/06/655006/3/gate/test-release-openstack/9b63479/ : SUCCESS in 3m 08s\n- tempest-full http://logs.openstack.org/06/655006/3/gate/tempest-full/7651f8b/ : SUCCESS in 1h 31m 20s\n- tempest-full-py3 http://logs.openstack.org/06/655006/3/gate/tempest-full-py3/b95a9e7/ : SUCCESS in 1h 18m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/655006/3/gate/openstack-tox-lower-constraints/c9ad4b6/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/06/655006/3/gate/openstack-tox-pep8/e503fce/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/06/655006/3/gate/openstack-tox-py27/af8bfa7/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/06/655006/3/gate/openstack-tox-py35/40bdbe0/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/06/655006/3/gate/openstack-tox-py36/5d2431c/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/06/655006/3/gate/openstack-tox-py37/d5e7785/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/06/655006/3/gate/openstack-tox-docs/d326702/html/ : SUCCESS in 3m 42s\n- build-openstack-releasenotes http://logs.openstack.org/06/655006/3/gate/build-openstack-releasenotes/95d24f6/html/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"237d32cadf1de9ade2509fdde361d1b5309bd840","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 20:45:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7acfb420ff0324e8419a9e29ff0fa101f9516da","revisions":{"1c81233a891fd5daa502d80ce4b1055756517472":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:38:04.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/06/655006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/06/655006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/06/655006/1"}}},"commit":{"parents":[{"commit":"25908e2a12452fde644f113c8bf6d0282f0c192b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/25908e2a12452fde644f113c8bf6d0282f0c192b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:37.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:38:00.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia0b66aac906b7920f246c4710ed06d63e11276d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1c81233a891fd5daa502d80ce4b1055756517472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1c81233a891fd5daa502d80ce4b1055756517472"}]},"branch":"refs/heads/master"},"4181d5d494eaf4f7805665979649df85f0a991d9":{"kind":"REWORK","_number":2,"created":"2019-04-23 14:02:48.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/06/655006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/06/655006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/06/655006/2"}}},"commit":{"parents":[{"commit":"25908e2a12452fde644f113c8bf6d0282f0c192b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/25908e2a12452fde644f113c8bf6d0282f0c192b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:37.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 14:02:39.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia0b66aac906b7920f246c4710ed06d63e11276d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4181d5d494eaf4f7805665979649df85f0a991d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4181d5d494eaf4f7805665979649df85f0a991d9"}]},"branch":"refs/heads/master"},"d7acfb420ff0324e8419a9e29ff0fa101f9516da":{"kind":"REWORK","_number":3,"created":"2019-05-05 06:45:57.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/06/655006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/06/655006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/06/655006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/06/655006/3"}}},"commit":{"parents":[{"commit":"25908e2a12452fde644f113c8bf6d0282f0c192b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/25908e2a12452fde644f113c8bf6d0282f0c192b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:37:37.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-05 06:45:44.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia0b66aac906b7920f246c4710ed06d63e11276d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d7acfb420ff0324e8419a9e29ff0fa101f9516da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d7acfb420ff0324e8419a9e29ff0fa101f9516da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
