)]}'
{"id":"openstack%2Fglance_store~657203","triplet_id":"openstack%2Fglance_store~master~I85ed8eb4f51a242580047d0dfe664ce070d63b6b","project":"openstack/glance_store","branch":"master","hashtags":[],"change_id":"I85ed8eb4f51a242580047d0dfe664ce070d63b6b","subject":"Fix failing tips-py35 jobs","status":"MERGED","created":"2019-05-05 06:30:04.000000000","updated":"2019-05-12 13:02:43.000000000","submitted":"2019-05-12 13:02:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657203-1557666163221-4a183efd","meta_rev_id":"8cac6a6452be82facdca7552b83280beb1ea91a7","_number":657203,"virtual_id_number":657203,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2019-05-12 09:19:10.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-05-12 13:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-05-09 09:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-05-10 12:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-05-10 12:29:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2019-05-09 09:39:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 09:39:29.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-05-10 12:29:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-12 13:02:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80d57cd6a3e40554ee0fc372db5e071a77782d52","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-05 06:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"144a1b11ded292e351c3f619540afa0c0de03ddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 08:22:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/03/657203/1/check/tempest-full/ec885ce/ : SUCCESS in 1h 51m 23s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/check/tempest-full-py3/0bf2c77/ : FAILURE in 26m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/check/openstack-tox-lower-constraints/71ac13d/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/check/openstack-tox-pep8/ca46932/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/check/openstack-tox-py27/1132874/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/check/openstack-tox-py35/d5031ae/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/check/openstack-tox-py36/72c3fe4/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/check/openstack-tox-py37/468f260/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/check/openstack-tox-docs/af437e9/html/ : SUCCESS in 4m 30s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/93333f6/ : FAILURE in 46m 43s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-sheepdog/87eedbb/ : FAILURE in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c837cc15664964ba96133535395c606a8972e7d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-05 09:00:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f707e0f5bfb2151ac80d87394c1e13eebd79dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 10:51:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/03/657203/1/check/tempest-full/0cd5167/ : SUCCESS in 1h 50m 02s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/check/tempest-full-py3/539d7ba/ : SUCCESS in 1h 28m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/check/openstack-tox-lower-constraints/734dd84/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/check/openstack-tox-pep8/aae4e19/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/check/openstack-tox-py27/47c76db/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/check/openstack-tox-py35/2901097/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/check/openstack-tox-py36/b5b1653/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/check/openstack-tox-py37/9ea5be7/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/check/openstack-tox-docs/d4e240d/html/ : SUCCESS in 3m 54s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/0a34f61/ : SUCCESS in 1h 20m 36s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-sheepdog/0c4f12c/ : FAILURE in 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba92a84d123a2df9e9e5f88c04546a9b01541b1d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-05-09 09:39:29.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nThanks Andreas,\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c73ab334d34a9df82e7eb8a3c61c85ede7584db0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-10 12:29:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, Andreas.\n\n@Erno: I wonder if we should do here what you did with https://review.opendev.org/#/c/657966/ and just make these use the py3 environment, and leave it up to infra to make sure py3 is pointing to the appropriate minor version?","accounts_in_message":[],"_revision_number":1},{"id":"5c01eef67f1b56c52168042aa01c94966f18386a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 12:29:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40b16cdcd5624feebf31d158cd75d2efdb39bcae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 14:34:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/03/657203/1/gate/tempest-full/1c75da5/ : SUCCESS in 2h 03m 13s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/gate/tempest-full-py3/270cb88/ : SUCCESS in 1h 30m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/gate/openstack-tox-lower-constraints/c146a98/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/gate/openstack-tox-pep8/9bb6bf0/ : FAILURE in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py27/eff9940/ : SUCCESS in 6m 12s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py35/78a13e7/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py36/e719b80/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py37/d7e7961/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/gate/openstack-tox-docs/e5789be/html/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a982caf0be1d9411ce6484c9241444ade981967c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-10 17:17:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71283527461ca5d8466a7879af19e537e1caf2aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:13:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/03/657203/1/check/tempest-full/45b0fe4/ : SUCCESS in 1h 52m 12s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/check/tempest-full-py3/4ac752c/ : FAILURE in 1h 53m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/check/openstack-tox-lower-constraints/db8c3a7/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/check/openstack-tox-pep8/0799481/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/check/openstack-tox-py27/21ced9a/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/check/openstack-tox-py35/e789876/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/check/openstack-tox-py36/d2cf7df/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/check/openstack-tox-py37/c8e7e0c/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/check/openstack-tox-docs/55ab1fb/html/ : SUCCESS in 4m 16s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/2240fed/ : SUCCESS in 1h 28m 19s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-sheepdog/d53b898/ : FAILURE in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdf63da364b50d7cb0f116f43071de37e4a7eea1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-10 19:16:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65ff63018cd35f489d9fda99df1804676e93d30d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:24:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/03/657203/1/check/tempest-full/e2feb79/ : SUCCESS in 1h 52m 06s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/check/tempest-full-py3/810cfe7/ : TIMED_OUT in 2h 06m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/check/openstack-tox-lower-constraints/7a19ac1/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/check/openstack-tox-pep8/dab1553/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/check/openstack-tox-py27/7de840d/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/check/openstack-tox-py35/9c27a54/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/check/openstack-tox-py36/db709fc/ : SUCCESS in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/check/openstack-tox-py37/36c0ab6/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/check/openstack-tox-docs/bcecd5a/html/ : SUCCESS in 3m 54s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/136b22e/ : SUCCESS in 1h 28m 15s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-sheepdog/61c4a80/ : FAILURE in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83cdbd801f5c6414fa4455df710a89d04ede009b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-11 15:25:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b243d1e4e6624ff7ae44f1825a8992fa3197aba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 17:13:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/03/657203/1/check/tempest-full/24eb45b/ : SUCCESS in 1h 43m 24s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/check/tempest-full-py3/cc18e66/ : SUCCESS in 1h 39m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/check/openstack-tox-lower-constraints/6ee08c2/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/check/openstack-tox-pep8/297dc70/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/check/openstack-tox-py27/64f0de7/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/check/openstack-tox-py35/5cca349/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/check/openstack-tox-py36/aa2dd19/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/check/openstack-tox-py37/b7c380c/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/check/openstack-tox-docs/3519966/html/ : SUCCESS in 3m 30s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/373ba0e/ : SUCCESS in 1h 25m 51s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-sheepdog/351734f/ : FAILURE in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dd0746b2698dbb4892bceb2eba05e388e559cc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 17:13:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8847865acbe25e504e140627af3cc74d2a18f6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 19:07:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/03/657203/1/gate/tempest-full/03d264d/ : FAILURE in 45m 53s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/gate/tempest-full-py3/ee79dad/ : SUCCESS in 1h 26m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/gate/openstack-tox-lower-constraints/dcfbce6/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/gate/openstack-tox-pep8/4a39e38/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py27/7d98d5c/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py35/9f3d455/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py36/3b310dd/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py37/f51e992/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/gate/openstack-tox-docs/97e55f5/html/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"da0c4d9f2401bce3580895e77b4a6fb4448d4197","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-12 09:19:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33a8bc0fee5fb60a04bccad40b583f65fc881ca1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 11:04:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/03/657203/1/check/tempest-full/c03e13a/ : SUCCESS in 1h 41m 37s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/check/tempest-full-py3/07e77af/ : SUCCESS in 1h 28m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/check/openstack-tox-lower-constraints/4dd4a23/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/check/openstack-tox-pep8/9fb2fc0/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/check/openstack-tox-py27/05de5e1/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/check/openstack-tox-py35/6da5f71/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/check/openstack-tox-py36/b1c6e36/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/check/openstack-tox-py37/5f36446/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/check/openstack-tox-docs/7592698/html/ : SUCCESS in 3m 40s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/91b232d/ : SUCCESS in 1h 22m 00s (non-voting)\n- legacy-tempest-dsvm-full-sheepdog http://logs.openstack.org/03/657203/1/check/legacy-tempest-dsvm-full-sheepdog/e9a83a6/ : FAILURE in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b94a29efe8472d903da8cca9d1de2a33cd750dd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 11:04:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2defdb998174df76a1aecef90216444659002799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 13:02:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/03/657203/1/gate/tempest-full/4910cde/ : SUCCESS in 1h 52m 48s\n- tempest-full-py3 http://logs.openstack.org/03/657203/1/gate/tempest-full-py3/4511142/ : SUCCESS in 1h 33m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/03/657203/1/gate/openstack-tox-lower-constraints/4450bd3/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/03/657203/1/gate/openstack-tox-pep8/cae758b/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py27/eed59d4/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py35/ae8c154/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py36/98d3f3e/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/03/657203/1/gate/openstack-tox-py37/8d122e5/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/03/657203/1/gate/openstack-tox-docs/52f63a4/html/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8cac6a6452be82facdca7552b83280beb1ea91a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 13:02:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fde1421bd3f71f2d9ccaea6c41c30f75864bb64","revisions":{"4fde1421bd3f71f2d9ccaea6c41c30f75864bb64":{"kind":"REWORK","_number":1,"created":"2019-05-05 06:30:04.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/657203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/03/657203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/657203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/657203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/657203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/03/657203/1"}}},"commit":{"parents":[{"commit":"25908e2a12452fde644f113c8bf6d0282f0c192b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/25908e2a12452fde644f113c8bf6d0282f0c192b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-05 06:28:03.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-05 06:30:01.000000000","tz":120},"subject":"Fix failing tips-py35 jobs","message":"Fix failing tips-py35 jobs\n\nWith Train, Python 3.5 is not a tested target anymore, switch to Python\n3.6. This should fix all failing tips-py35 jobs.\n\nExample failure:\nhttp://logs.openstack.org/periodic/opendev.org/openstack/glance_store/master/glance_store-tox-py35-cinder-tips/d256a41/\n\nChange-Id: I85ed8eb4f51a242580047d0dfe664ce070d63b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4fde1421bd3f71f2d9ccaea6c41c30f75864bb64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4fde1421bd3f71f2d9ccaea6c41c30f75864bb64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
