)]}'
{"id":"openstack%2Fglance_store~660450","triplet_id":"openstack%2Fglance_store~master~I79d9dbc8f6ebea381d99e96c4ba4194ba0188b8a","project":"openstack/glance_store","branch":"master","topic":"sheepdog","hashtags":[],"change_id":"I79d9dbc8f6ebea381d99e96c4ba4194ba0188b8a","subject":"Remove sheepdog tests from zuul config","status":"MERGED","created":"2019-05-21 16:26:21.000000000","updated":"2019-07-05 13:17:16.000000000","submitted":"2019-07-05 13:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660450-1562332636455-31ef7618","meta_rev_id":"d4498f4bb64d4c5a7b57c31d9579c222f9cf2875","_number":660450,"virtual_id_number":660450,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-07-05 13:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-07-05 07:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-06-24 14:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-06-26 05:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2019-07-05 07:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-06-26 05:53:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 14:28:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-26 05:53:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-07-05 07:28:57.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-07-05 13:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5af6a961722edabbfe760f5c1a34c3cbaa3351d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-21 16:26:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"516ede9d8576667f838c41b3f0fea5d62a319a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 18:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/50/660450/1/check/tempest-full/673dcfd/ : SUCCESS in 1h 44m 36s\n- tempest-full-py3 http://logs.openstack.org/50/660450/1/check/tempest-full-py3/df9e7ff/ : SUCCESS in 1h 27m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/660450/1/check/openstack-tox-lower-constraints/a22e1eb/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/50/660450/1/check/openstack-tox-pep8/ddb5c62/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/50/660450/1/check/openstack-tox-py27/32796bf/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/50/660450/1/check/openstack-tox-py35/d701c00/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/50/660450/1/check/openstack-tox-py36/0426cc0/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/50/660450/1/check/openstack-tox-py37/04e80d4/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/50/660450/1/check/openstack-tox-docs/01d4568/html/ : SUCCESS in 5m 24s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store http://logs.openstack.org/50/660450/1/check/legacy-tempest-dsvm-full-ceph-plugin-src-glance_store/5d1b2d6/ : SUCCESS in 1h 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a033ff94e06796fffccb8623bcb34eae8caa435","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-21 20:19:49.000000000","message":"Patch Set 1: Code-Review+1\n\nthis is needed for removing sheepdog and fine, thanks.\n\nNeeds backporting to all open branches.","accounts_in_message":[],"_revision_number":1},{"id":"1c221217d52d2284c83c4f03737cab64fe03b38a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-06-19 11:08:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f76acf50f155df5ad901d865e5473a9ede87fd1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-24 14:27:11.000000000","message":"Patch Set 1:\n\nlet me rebase this...","accounts_in_message":[],"_revision_number":1},{"id":"145c4abcf60ae8dd690682b03106f40b3a23c24b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-24 14:28:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"464369820c7655351f5512bc098784dc8686627c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-24 14:28:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cf72ed277291cff7115183085dfdfca41db953e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 16:27:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/50/660450/2/check/tempest-full/a2df8b9/ : SUCCESS in 1h 56m 29s\n- tempest-full-py3 http://logs.openstack.org/50/660450/2/check/tempest-full-py3/3387af3/ : SUCCESS in 1h 50m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/660450/2/check/openstack-tox-lower-constraints/c418779/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/50/660450/2/check/openstack-tox-pep8/dde9d54/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/50/660450/2/check/openstack-tox-py27/09fcc8f/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/50/660450/2/check/openstack-tox-py36/9c1c97f/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/50/660450/2/check/openstack-tox-py37/78ed48b/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/50/660450/2/check/openstack-tox-docs/ffd807e/html/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"4dd08f20af40e302f0d13e0149905837b0887c20","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-06-26 05:53:01.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"f9895890400408844611219dd765c7e7a2e24494","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-05 07:28:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d62a932e09b1b14cdb6c54f57e390e05692027c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 07:29:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33aab595dcbaccf89971cec660226b30924ed478","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 09:08:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/50/660450/2/gate/tempest-full/8580fa0/ : FAILURE in 1h 35m 45s\n- tempest-full-py3 http://logs.openstack.org/50/660450/2/gate/tempest-full-py3/010cf3a/ : SUCCESS in 1h 30m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/660450/2/gate/openstack-tox-lower-constraints/cd5b2d3/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/50/660450/2/gate/openstack-tox-pep8/35717ee/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/50/660450/2/gate/openstack-tox-py27/3ffb379/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/50/660450/2/gate/openstack-tox-py36/3bacbd7/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/50/660450/2/gate/openstack-tox-py37/3843e87/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/50/660450/2/gate/openstack-tox-docs/6a4961a/html/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3719d795e694fa86f6d08354ea28c84db0bc0031","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-05 09:39:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a199cd10e4fe303bdce443dfde294e541bec5b16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 11:25:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/50/660450/2/check/tempest-full/dcbaf02/ : SUCCESS in 1h 40m 01s\n- tempest-full-py3 http://logs.openstack.org/50/660450/2/check/tempest-full-py3/b39904d/ : SUCCESS in 1h 41m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/660450/2/check/openstack-tox-lower-constraints/6bd0a6a/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/50/660450/2/check/openstack-tox-pep8/8a4e543/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/50/660450/2/check/openstack-tox-py27/3e0cfc3/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/50/660450/2/check/openstack-tox-py36/25d6acc/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/50/660450/2/check/openstack-tox-py37/e3fc8c7/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/50/660450/2/check/openstack-tox-docs/082d9e0/html/ : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"92b383a67fcfcda476e3db1f7f44e8a1da88d5f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 11:25:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"927e96f624e9dea863d47c8d01ea1e8b7bfbf372","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:17:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a06ebfe792ad015c739867b28894e2646bc6a73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:17:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/50/660450/2/gate/tempest-full/e4cdad5/ : SUCCESS in 1h 49m 38s\n- tempest-full-py3 http://logs.openstack.org/50/660450/2/gate/tempest-full-py3/4b59c5c/ : SUCCESS in 1h 23m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/660450/2/gate/openstack-tox-lower-constraints/5dc3711/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 http://logs.openstack.org/50/660450/2/gate/openstack-tox-pep8/ac70da5/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/50/660450/2/gate/openstack-tox-py27/b843eaa/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/50/660450/2/gate/openstack-tox-py36/78da183/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/50/660450/2/gate/openstack-tox-py37/dfe5bdc/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/50/660450/2/gate/openstack-tox-docs/62fa0fc/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7a6064e7dc4775d08753f029c2589a4586e0f66","revisions":{"1ad859111149c21c0393c112224596e95c112133":{"kind":"REWORK","_number":1,"created":"2019-05-21 16:26:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/50/660450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/50/660450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/50/660450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/50/660450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/50/660450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/50/660450/1"}}},"commit":{"parents":[{"commit":"4d7898a345d90ee3b1d622a0767b4d9300265eda","subject":"Merge \"Python3: Fix return type on CooperativeReader.read\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/4d7898a345d90ee3b1d622a0767b4d9300265eda"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-21 16:24:18.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-21 16:24:18.000000000","tz":-300},"subject":"Remove sheepdog tests from zuul config","message":"Remove sheepdog tests from zuul config\n\nThe sheepdog project is no longer maintained [0], so this removes the\ntest configuration as part of an effort to clean up those test\ndefinitions. The glance_store support for sheepdog should also be\ndeprecated and removed, but that will be done separately.\n\n[0] http://lists.wpkg.org/pipermail/sheepdog/2019-March/068451.html\n\nChange-Id: I79d9dbc8f6ebea381d99e96c4ba4194ba0188b8a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1ad859111149c21c0393c112224596e95c112133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/1ad859111149c21c0393c112224596e95c112133"}]},"branch":"refs/heads/master"},"a7a6064e7dc4775d08753f029c2589a4586e0f66":{"kind":"REWORK","_number":2,"created":"2019-06-24 14:28:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/660450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/50/660450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/50/660450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/50/660450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/50/660450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/50/660450/2"}}},"commit":{"parents":[{"commit":"a11649d4df5655329b47c02accd35ed95c948633","subject":"Add 0.29.1 releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a11649d4df5655329b47c02accd35ed95c948633"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-21 16:24:18.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-24 14:28:08.000000000","tz":120},"subject":"Remove sheepdog tests from zuul config","message":"Remove sheepdog tests from zuul config\n\nThe sheepdog project is no longer maintained [0], so this removes the\ntest configuration as part of an effort to clean up those test\ndefinitions. The glance_store support for sheepdog should also be\ndeprecated and removed, but that will be done separately.\n\n[0] http://lists.wpkg.org/pipermail/sheepdog/2019-March/068451.html\n\nChange-Id: I79d9dbc8f6ebea381d99e96c4ba4194ba0188b8a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a7a6064e7dc4775d08753f029c2589a4586e0f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a7a6064e7dc4775d08753f029c2589a4586e0f66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
