)]}'
{"id":"openstack%2Fglance_store~686918","triplet_id":"openstack%2Fglance_store~master~I267224f734303e4e9791e5019a014814f6161956","project":"openstack/glance_store","branch":"master","hashtags":[],"change_id":"I267224f734303e4e9791e5019a014814f6161956","subject":"Add release notes link in readme","status":"MERGED","created":"2019-10-06 06:37:27.000000000","updated":"2019-10-09 13:41:07.000000000","submitted":"2019-10-09 13:38:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686918-1570628333214-14372de2","meta_rev_id":"33ce29a3dc873576192195a2e6db54845c9dc93e","_number":686918,"virtual_id_number":686918,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 13:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-10-09 13:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2019-10-09 13:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 13:28:33.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-10-09 13:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a503f343636237088ab69c989a47e8950016226","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-06 06:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c644d55f9d620492953eae28b525b4ef14a9882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 06:43:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e9710d502e347b6b7b362b8cbee4a6c : SUCCESS in 1m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d173998b03844d6b618e5b59f054b08 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4056d8cc184049efbb884aa158a17ded : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"eeb5610e344d9322e9ef6547064c0502d2159383","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-09 13:28:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"e174f8e843a06177fb78ed4f00da8b314f33325d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 13:28:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d47753183ee64cacbb9ceb6ca5566276931e4f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 13:38:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3e2a3a7e9554f95a67d0913a5bb8236 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baba28b2c20e43dbaeec8d7c8a7d5b5b : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0203204c5d614ee1ad75045efb9ab535 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6396892d5d9485d626a36ae2cc9dc85721b0d959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 13:38:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33ce29a3dc873576192195a2e6db54845c9dc93e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 13:41:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/379c8692277142a78c20400b642fef5e : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d68c69fba7046db81a10593d18935481fc7ef5be","revisions":{"d68c69fba7046db81a10593d18935481fc7ef5be":{"kind":"REWORK","_number":1,"created":"2019-10-06 06:37:27.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/18/686918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/18/686918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/686918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/686918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/18/686918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/18/686918/1"}}},"commit":{"parents":[{"commit":"ae249603a5c8511ba166391ad6051c8b7a544f3d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ae249603a5c8511ba166391ad6051c8b7a544f3d"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-06 06:37:10.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-06 06:37:10.000000000","tz":480},"subject":"Add release notes link in readme","message":"Add release notes link in readme\n\nChange-Id: I267224f734303e4e9791e5019a014814f6161956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d68c69fba7046db81a10593d18935481fc7ef5be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d68c69fba7046db81a10593d18935481fc7ef5be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
