)]}'
{"id":"openstack%2Fglance_store~694903","triplet_id":"openstack%2Fglance_store~master~I42da1cb2e2bd626bf0b99c747f187b6f33674810","project":"openstack/glance_store","branch":"master","topic":"test","hashtags":[],"change_id":"I42da1cb2e2bd626bf0b99c747f187b6f33674810","subject":"Do Not Merge!!!","status":"ABANDONED","created":"2019-11-19 05:38:22.000000000","updated":"2019-11-22 07:09:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"856f0977d7b7a3bf4da18fb90bf1639554b106cd","_number":694903,"virtual_id_number":694903,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-19 07:12:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2019-11-19 05:47:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 07:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ee5ac5596946b035a28877d55eaac0542fbd911","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-19 05:38:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f133c4b2186dd37db65dfe395ac55171f1d39546","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-19 05:47:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4b202be66be9df4165c708eefd9c4285bf43c6ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 07:12:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f219490174b54af8a55bef134c8f89dc : FAILURE in 18m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2db31db91e494fa1ba51acedfac1af2e : SUCCESS in 1h 33m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a302ba7e0fe4296b48626b836c3edeb : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a4c2a31c7d4257a9803464510392c1 : SUCCESS in 2m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed23052aa1cc4908a3c5ee4d805f6f47 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16d6f6542069449aa814ecb52021f70e : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b148f3cdf31840e2badb8d0c74296ac1 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39314fbd1179415da1e012c6d55a855e : SUCCESS in 3m 16s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/18943071fb934d83a862605b2b413019 : FAILURE in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d280d37b51cbe432b0da7a018c993a7b522e6c4d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-22 07:09:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb7093fcff1bf3e449912db710d2d9a238ce7dc8","revisions":{"eb7093fcff1bf3e449912db710d2d9a238ce7dc8":{"kind":"REWORK","_number":1,"created":"2019-11-19 05:38:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/694903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/03/694903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/694903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/694903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/03/694903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/03/694903/1"}}},"commit":{"parents":[{"commit":"d68c69fba7046db81a10593d18935481fc7ef5be","subject":"Add release notes link in readme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d68c69fba7046db81a10593d18935481fc7ef5be"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-19 05:37:58.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-19 05:38:22.000000000","tz":0},"subject":"Do Not Merge!!!","message":"Do Not Merge!!!\n\nChange-Id: I42da1cb2e2bd626bf0b99c747f187b6f33674810\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/eb7093fcff1bf3e449912db710d2d9a238ce7dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/eb7093fcff1bf3e449912db710d2d9a238ce7dc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
