)]}'
{"id":"openstack%2Fglance_store~709048","triplet_id":"openstack%2Fglance_store~stable%2Ftrain~I07064f3cb1a33ac4ac7e4b572f8e1b5c688bd9a3","project":"openstack/glance_store","branch":"stable/train","topic":"bug/1863983-stable/train","hashtags":[],"change_id":"I07064f3cb1a33ac4ac7e4b572f8e1b5c688bd9a3","subject":"Image upload fails if cinder multipath is enabled","status":"MERGED","created":"2020-02-21 05:10:09.000000000","updated":"2020-05-08 09:31:54.000000000","submitted":"2020-05-08 09:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709048-1588930209641-23512575","meta_rev_id":"f1755bf2f0f2f9df4d1c94a9282a2b8e5804fe4c","_number":709048,"virtual_id_number":709048,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-05-08 07:29:50.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 09:30:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 12:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-05 12:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-05-08 07:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-06 13:59:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-05-08 07:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-05-05 17:17:43.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 12:28:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-05 12:53:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-05-05 17:17:43.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2020-05-06 13:59:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-05-08 07:35:28.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-05-08 09:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c736447a6fa4b66ef583abbc87647072996c8356","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-21 05:10:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6b65508829cea7998025c1f33c284451c6d5effc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 07:06:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/208906d5479942a6a20f96617b964f98 : SUCCESS in 1h 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4f8be2a528b476783ca462336e9c33f : SUCCESS in 1h 20m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/282e449192ca436baa77ab1fecd5fc9d : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffde19d83d64c0e87259cf5f1266cd3 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0183f2389b06474f80358f94f5d0dde8 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e24da1c38447ecb209dfcf08c18a23 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d63e6795ba5b4dafa25082cc57e68584 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94af2252a4b041f8b15d16ef3cdc4969 : SUCCESS in 4m 00s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/6856f6893b864e0c99019edfcaaa6ef8 : SUCCESS in 1h 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0187dde49512f25851404f725029f4bceec15b6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 12:28:56.000000000","message":"Patch Set 1: Code-Review+2\n\nGenerally we try to avoid backporting the addition of new config options, but it looks like the defaults are safe so that there won\u0027t be any unexpected behavior changes after upgrading.","accounts_in_message":[],"_revision_number":1},{"id":"050ede6fc8e3a5ce95349cd2d9a9148adbea06bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-05 13:49:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"540cd8f687f21d75a817c862516b8a990a24a5f0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-05 05:26:15.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a1da8f19a8c8a14ac5bf4e8c73f1c641f5bbe9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 07:17:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/893d7740e0b2476e8d09d625f5aea372 : SUCCESS in 1h 49m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7621586618fa4b1786a2764eab3d5674 : SUCCESS in 1h 19m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28d14954388842938a19624f1c7455f6 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24be4e16be374ae5844b2dce0a106d4d : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dea9da5f3f324306b655942f8c2f8c6b : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cad4a8a0c8ae4665b5b22520f4f8d437 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b2d05fa4cf34bf6b1d1cee03672a3c7 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6f3e944c3f41ffb7fc0274b2a51671 : SUCCESS in 4m 09s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/8127c687409f4f3894115c636e24a1d5 : SUCCESS in 1h 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fc5353a6d5a9158fd137b7a66b3b3c098ff9a8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-05-05 12:53:45.000000000","message":"Patch Set 1: Code-Review+2\n\nAgree with Sean that the defaults look safe.  Multipath is a pretty popular configuration for running Cinder, so I think this is worth backporting.  Plus, the actual code change is very small, even though it does add two config options.\n\nNot hitting +W to give other glance-stable-maint cores time to express an opinion.","accounts_in_message":[],"_revision_number":1},{"id":"90b145d2cfd7665b476088570a5138fc7e15ee4d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-05-06 13:59:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"02c98a141c68ebfd46286b2c11a149eb319508d8","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-05-08 07:21:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4eb800b0851db2a0c2a2c587689da9b7c0319fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-08 07:29:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b81c0e9b7e5ecf9b4f87c1b130d8844f5679ae7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-05-08 07:35:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c27bd32d147be65b621ff6daf165272f7e0e46ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 07:35:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7c23dad5d22011a108ce45b0feef2cee70f4fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:30:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d7857a844ef74a5790b607a85e422050 : SUCCESS in 1h 52m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79a329ae4aad47f0af8cc7dbdb8c61ac : SUCCESS in 1h 41m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88e3763f90a34f26b625971269ba13c4 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939f625507b642c48183022ab4145393 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef56250ac2674be58c9afc667f70c8c8 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52aa9c06fb2d414d91f388efae7b88fa : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d83a605d5a741fbaa14fa7a8d1e928d : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa2f930523c40af929ada11600c0bdf : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"508101fbbca147f2d3e5febb63beaa391f6c7c14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:30:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ea83a6cdece46b9cc26569b5d0d562af9f36e70b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:31:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ca656cea484f08ad4d1b5614cf13f8 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d052ce57b5cfc66a4b3b8030a14db3d46c312990","revisions":{"d052ce57b5cfc66a4b3b8030a14db3d46c312990":{"kind":"REWORK","_number":1,"created":"2020-02-21 05:10:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/48/709048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/48/709048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/709048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/709048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/709048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/48/709048/1"}}},"commit":{"parents":[{"commit":"9383fde396041758d767ce3f30b9a5669ad9ebf1","subject":"Merge \"Release note for 1.0.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9383fde396041758d767ce3f30b9a5669ad9ebf1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-20 07:21:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-21 05:10:09.000000000","tz":0},"subject":"Image upload fails if cinder multipath is enabled","message":"Image upload fails if cinder multipath is enabled\n\nImage upload is failing with NoFibreChannelVolumeDeviceFound error after\nconfiguring cinder backend (HP3Par FC storage) for the glance service. The\nissue here is that usually cinder calls os_brick with multipath information\nbut in this case cinder driver of glance is doing the call (without passing\nany multipath information). with hard-coded\u0027 False to let os_brick library\nknow that multipath is not configured.\n\nIn order to fix this, added two new boolean config option for cinder\ndriver default to False;\n\n\u0027cinder_use_multipath\u0027\n\u0027cinder_enforce_multipath\u0027\n\nChange-Id: I07064f3cb1a33ac4ac7e4b572f8e1b5c688bd9a3\nCloses-Bug: #1863983\n(cherry picked from commit ae73287c7b4db8d5894f72d8a5b1b89b42697545)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d052ce57b5cfc66a4b3b8030a14db3d46c312990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d052ce57b5cfc66a4b3b8030a14db3d46c312990"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
