)]}'
{"id":"openstack%2Fglance_store~709049","triplet_id":"openstack%2Fglance_store~stable%2Fstein~I07064f3cb1a33ac4ac7e4b572f8e1b5c688bd9a3","project":"openstack/glance_store","branch":"stable/stein","topic":"bug/1863983-stable/stein","hashtags":[],"change_id":"I07064f3cb1a33ac4ac7e4b572f8e1b5c688bd9a3","subject":"Image upload fails if cinder multipath is enabled","status":"ABANDONED","created":"2020-02-21 05:11:14.000000000","updated":"2020-03-05 13:49:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"569d4b41834281d150840e3e631b50b654d724f7","_number":709049,"virtual_id_number":709049,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"date":"2020-02-21 08:50:13.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-21 10:30:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-02-21 05:11:36.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-21 05:11:27.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 05:11:27.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-21 05:11:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-21 10:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3eb2e80dd953c903e3fe6bf67c09a8f9c18bca0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-21 05:11:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f51b47267a3ceb2814e88263ecaf36eb947fb68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 06:42:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/537ff051d06a435e9ed0e0c9d9f440ac : SUCCESS in 1h 30m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6afc74dcf9ef40138974a45c7520de7f : FAILURE in 1h 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90b27da63b3946308ca90b01c89fbe61 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870d32e3964c49599e4511bba42984ba : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3b533aa0663412690f3682a51c0f4d8 : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55d6bc7b9d7f4c8ba618206ce712b2a5 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0978632a9240f3aa923a6d655fae50 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3a8ab263e854c299940cb03a00c6091 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58577f1e11204692aa413acc5ef8a281 : SUCCESS in 4m 37s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/1f7ded7e540c470a8ef957ae90f57df2 : FAILURE in 1h 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7958f1877024909e339890a626af824f50ee9673","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-21 06:49:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e18696e839cfb22c735f6e4a7ee06a8ae960ca3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 08:45:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d657c82346e048ec9799ce158e7ecb31 : FAILURE in 1h 53m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25cbe5dacaf64da1b263d20edac6fe1c : FAILURE in 1h 41m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b4a19b7ffc145e98b4083373e42aa58 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300c59ba2595481ab7375e4cbc1c0397 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22b410d9b70349d1abd61d398fc7f31a : SUCCESS in 6m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d9c5e3244a2c44dabb657e3a6a663768 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e37f39c085048af8ff280e054596ded : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26c742c69499499cacffd3627009d59e : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5401e54eda243f28a7b6b9416cfd7a1 : SUCCESS in 5m 37s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/906e7d5e8db64757a96c6bdfac00c15d : SUCCESS in 1h 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"615fcefec288b5fb69273c412810e2574f08b13d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-21 08:50:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba15c589fa66bb6c40647efd5a14e216e64b91da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 10:30:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/67c535c1281c479ebad4a6c5eba8ebdd : SUCCESS in 1h 30m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33d37277fc9d4742aad247b0f7cc2fd0 : SUCCESS in 1h 35m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5aefe1dbe2fd405bad88576ff21d62f1 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32cc118999bd41feb98f8549c774fc59 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfc2b0b82bac4d5384300f9cd1ac0c81 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b638fb560a3d414184a89a06c2b211d9 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa48d891b4c487f822d91e708d6c286 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79786d3ec65d4b478446eef800facc28 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0681707ff3fe4a11a7ce5763e59b11b0 : SUCCESS in 4m 19s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/4db5ebfbab314f5faae4afae85b2941f : SUCCESS in 1h 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50fe265003ea035b85a6b7975073998b6621ee41","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-05 13:49:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4a43eee4bd4943a78d2e32dc17bf94444dbbe23","revisions":{"c4a43eee4bd4943a78d2e32dc17bf94444dbbe23":{"kind":"REWORK","_number":1,"created":"2020-02-21 05:11:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/49/709049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/49/709049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/709049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/709049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/49/709049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/49/709049/1"}}},"commit":{"parents":[{"commit":"27e421af26cd1bcc68746a0beadb541fb8d647be","subject":"Remove sheepdog tests from zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/27e421af26cd1bcc68746a0beadb541fb8d647be"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-20 07:21:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-21 05:11:14.000000000","tz":0},"subject":"Image upload fails if cinder multipath is enabled","message":"Image upload fails if cinder multipath is enabled\n\nImage upload is failing with NoFibreChannelVolumeDeviceFound error after\nconfiguring cinder backend (HP3Par FC storage) for the glance service. The\nissue here is that usually cinder calls os_brick with multipath information\nbut in this case cinder driver of glance is doing the call (without passing\nany multipath information). with hard-coded\u0027 False to let os_brick library\nknow that multipath is not configured.\n\nIn order to fix this, added two new boolean config option for cinder\ndriver default to False;\n\n\u0027cinder_use_multipath\u0027\n\u0027cinder_enforce_multipath\u0027\n\nChange-Id: I07064f3cb1a33ac4ac7e4b572f8e1b5c688bd9a3\nCloses-Bug: #1863983\n(cherry picked from commit ae73287c7b4db8d5894f72d8a5b1b89b42697545)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c4a43eee4bd4943a78d2e32dc17bf94444dbbe23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c4a43eee4bd4943a78d2e32dc17bf94444dbbe23"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
