)]}'
{"id":"openstack%2Fglance_store~716874","triplet_id":"openstack%2Fglance_store~master~I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6","project":"openstack/glance_store","branch":"master","topic":"bug/1870289","hashtags":[],"change_id":"I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6","subject":"Add lock per share for cinder nfs mount/umount","status":"MERGED","created":"2020-04-02 07:27:12.000000000","updated":"2020-05-01 15:32:35.000000000","submitted":"2020-05-01 15:31:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":81,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716874-1588347065094-2ccb4115","meta_rev_id":"be770f12d489c5c29bf651742520138413fbf728","_number":716874,"virtual_id_number":716874,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-01 15:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-01 13:33:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-05-01 12:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-01 13:33:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 14:41:01.000000000","updated_by":{"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},"reviewer":{"_account_id":22700,"name":"Paul Grist","email":"pgrist@redhat.com"},"state":"REVIEWER"},{"updated":"2020-04-28 13:46:33.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-04-29 18:19:19.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2020-04-29 18:27:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-01 12:45:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-01 13:33:26.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-05-01 15:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b8a660d0111f62256ba9a4904e94d2702f674e9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-02 07:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee755e7ee34a41ba48ed41e43d55d310e9349d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 08:53:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1978b85ef554e63bb114a82cc6269fe : SUCCESS in 1h 25m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a25ff4bf7eb43019c020c2ea78d20d5 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3522c1ed1140b9a0afa34fda3defa1 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7510297ca1214208908c28ece39ef852 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14463e4919814e7a98af4db28a2e5729 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8cc1afa83d544f58dba6242d74efb92 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0eb35bffc8641298cdb932205ae13d0 : SUCCESS in 3m 52s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/0be115089a54417e883e904b90857af8 : SUCCESS in 1h 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd7fdd06dd12584a9aba9d78b5c0237a9e528547","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-02 13:51:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"806e35a01e582cb40d99cc5db2d276a3b62c987e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:15:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26d53fc5fe86437483f26e9b08e4d5fa : SUCCESS in 1h 35m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/159183fa70db4cdb8cb5cff1b130538c : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f9120a66464f9f90730abe05a93349 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b624b8f243424cbab3c625544eaebe6a : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e65e53b6262b4dda929ecde413da8c76 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6281ee8a12be4bc5a407caf55b585866 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258cb87615b64f479d6898bf9bbded0b : SUCCESS in 7m 06s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/799dd3a12a324e5584e1eb119b5d068b : SUCCESS in 1h 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"562fa237594b721acd63b35ee0a3995ce1cc7439","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-13 13:57:10.000000000","message":"Patch Set 2:\n\nChange looks good, but we need unit test for the coverage.","accounts_in_message":[],"_revision_number":2},{"id":"787a89e1390382e9ab18253dafb3c2e40d2c863f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-17 12:03:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"87a42639c6c5b7a1c17c7c787143afeba18f1330","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-17 12:59:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc7fc98c542ddc974ad53aa5ab685d4a0257c5a8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-17 13:07:26.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nLooks good. Some comments that are not critical at all.\n\nAgree with Abhishek though that it would be good to have unit test coverage.","accounts_in_message":[],"_revision_number":4},{"id":"fd0e4d3345c07f5b4dd3ca96d6b43a8560c23201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 14:41:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e41970cb982c4c85a5836cb868013ca1 : SUCCESS in 1h 27m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aac98cf22b64a7da40d264e499ffcde : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659bbda1db2046d59bddd2d4a0789ca8 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f00863492144f299f7f1e97ab321c52 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be0f9401d000498bbfb18bc10e229c3a : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903ebffdb53740fdaa76e7117fca1791 : SUCCESS in 4m 26s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/0a7b1674f79143e09bf7ff7d420764fe : SUCCESS in 1h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb046f483fb71d6906e29a1d5f163b38ebace347","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-17 15:41:02.000000000","message":"Patch Set 4: Code-Review-2\n\n(10 comments)\n\nTo avoid any confusion it seems like cinder driver needs to move under it\u0027s own subfolder glance_store/_drivers/cinder/ similar manner as swift is.\n\nThis seems to be memory stated only. What happens if the service crashes and restarts without cleaning the mounts in between?\n\nBunch of other comments inline.\n\n-2 for the Singleton approach/implementation and lack of subfoldering once those are reworked ping me to lift it.","accounts_in_message":[],"_revision_number":4},{"id":"52a0873be7c0294fd0bd114d45be95947a4d89e4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-21 18:20:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nIMO some concerns raised by Erno are valid and we need to work around it, specially using of singleton class and retry logic.","accounts_in_message":[],"_revision_number":4},{"id":"b5f39afada2abff3d1859f54223712e867641517","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-21 20:08:10.000000000","message":"Patch Set 4:\n\n(4 comments)\n\n\u003e (10 comments)\n \u003e \n \u003e To avoid any confusion it seems like cinder driver needs to move\n \u003e under it\u0027s own subfolder glance_store/_drivers/cinder/ similar\n \u003e manner as swift is.\n \u003e \n \u003e This seems to be memory stated only. What happens if the service\n \u003e crashes and restarts without cleaning the mounts in between?\n \u003e \n \u003e Bunch of other comments inline.\n \u003e \n \u003e -2 for the Singleton approach/implementation and lack of\n \u003e subfoldering once those are reworked ping me to lift it.\n\nHi Erno,\nI\u0027ve answered the singleton part inline.\nI will do the subfoldering.\nI had the same thought while implementing and the best way i could think of was to cleanup when the host comes up but currently glance doesn\u0027t store the service state in DB hence no way of knowing that.\nCould you suggest any other method to perform the same.","accounts_in_message":[],"_revision_number":4},{"id":"bd879f0c3bc0ceca5e2e483dfc5c84b4d60b5689","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-22 10:24:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d3aa649fbd518b8c0b1af7120142ba167daa48e8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-22 10:25:24.000000000","message":"Patch Set 5:\n\n(10 comments)\n\nThanks for the review Erno, Abhishek and Sean.","accounts_in_message":[],"_revision_number":5},{"id":"1b82a120616903d1bc1c8a5120915ee6c2cc073f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-22 11:34:06.000000000","message":"Patch Set 5:\n\nAlso just for additional reference, nova has the same singleton implementation https://review.opendev.org/#/c/383859/","accounts_in_message":[],"_revision_number":5},{"id":"0c03740344a01a96b2b00919d8a2e56e050d5d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:10:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3a22d5f26284a739aebd0693a50bb93 : SUCCESS in 1h 44m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93b69a72892146588b93a4f3e7e261a1 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d575607bf4943c28c0fb3a2a8230007 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a693cf03e24800b7819d4af8dc4e09 : FAILURE in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d257fddadcb546d6b61b335df4388611 : FAILURE in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa54f90f5174e01add8b1bd373ea68f : SUCCESS in 5m 23s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/e7cff23ab72e471d98cbed14bcd0c9b3 : SUCCESS in 1h 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5cea51c74ddc35a863f1f8065d9881724803fba3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-22 13:48:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5e5ecc81368d9e3440314fb88881674e105f946b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-22 13:55:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b46c51423659744ff111818c5f264976bde7329d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-04-22 14:18:57.000000000","message":"Patch Set 7: Code-Review-1\n\nSurely the bugfix here and work to move the cinder driver into a separate directory should be separate patches.","accounts_in_message":[],"_revision_number":7},{"id":"e7764e12440e914d064039af704d33415f54fa48","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-22 14:34:22.000000000","message":"Patch Set 7:\n\n\u003e Surely the bugfix here and work to move the cinder driver into a\n \u003e separate directory should be separate patches.\n\nNope, the fs_mount.py (integral part of that bigfix) is the reason why we need to directorize this. The _drivers is single entrypoint path for those drivers. So putting the cinder driver with it\u0027s fs_mount module into it\u0027s own directory prevents anyone getting confused fs_mount being it\u0027s own driver or part of some other drivers in the path.","accounts_in_message":[],"_revision_number":7},{"id":"51e3e82e3a6c35305b5889ea540af80911d242b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 15:30:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9f851488a9846539f5e711143512e7f : SUCCESS in 1h 20m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e5a911b204947959de732c6aee8f15b : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d293d2fe40406690f7c7e801781147 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d321aed402748449a81024b4963b912 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9cfdbbb89d749ab8edad7695a536499 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503b2b111a1248e6803d7c5b134f0740 : SUCCESS in 5m 07s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/67f9ad375db644368b7f973e6c192279 : SUCCESS in 1h 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f2d8cb80b0dc417856217b1834167663851d30b8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-23 09:32:17.000000000","message":"Patch Set 7: Review-Priority+1\n\n\u003e \u003e Surely the bugfix here and work to move the cinder driver into a\n \u003e \u003e separate directory should be separate patches.\n \u003e \n \u003e Nope, the fs_mount.py (integral part of that bigfix) is the reason\n \u003e why we need to directorize this. The _drivers is single entrypoint\n \u003e path for those drivers. So putting the cinder driver with it\u0027s\n \u003e fs_mount module into it\u0027s own directory prevents anyone getting\n \u003e confused fs_mount being it\u0027s own driver or part of some other\n \u003e drivers in the path.\n\nConsidering the scenario both opinions are correct. So to avoid this restructuring at the moment I have couple of suggestions.\n\n1. Move contents of new file \u0027fs_mount.py\u0027 under cinder.py and refactor it in Victoria cycle\n2. Create new module under glance_store/common as \u0027fs_utils.py\u0027\n\nKindly suggest!!","accounts_in_message":[],"_revision_number":7},{"id":"0fe44fed4575b426a6a37a82f7485c1222ebe387","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-24 09:00:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6840f151578b3918804ac7d1821dd3fbbd74d54d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:28:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99373b8d4f68448d9ef1544a6f129f16 : SUCCESS in 1h 24m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3b27bcf3d1e45d684946e5dc19c7ae9 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449fcaf6b3934fbdabb27510a8a715de : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce1cfebdfdad4e7994960881a5423036 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/799b6660171a470898ef194db3181086 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf65c671571c4604901c07506b04032e : FAILURE in 4m 58s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/1061a135b62a4c248b1690a8a6763f92 : SUCCESS in 1h 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"84de71d56d6b90a12bbb110118d5ff7780cf85bf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-27 07:57:59.000000000","message":"Patch Set 8:\n\nI tried the following code with metaclass[1] with single worker but it created new manager everytime[2].\nThe same thing works with my initial code producing the same manager.\n\n[1] http://paste.openstack.org/show/792739/\n[2] http://paste.openstack.org/show/792740/","accounts_in_message":[],"_revision_number":8},{"id":"4df976fac312a6193e7acc6760b46b0149e96ab3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-27 13:22:35.000000000","message":"Patch Set 8:\n\n\u003e I tried the following code with metaclass[1] with single worker but\n \u003e it created new manager everytime[2].\n \u003e The same thing works with my initial code producing the same\n \u003e manager.\n \u003e \n \u003e [1] http://paste.openstack.org/show/792739/\n \u003e [2] http://paste.openstack.org/show/792740/\n\nWhen you create the manager class, you need to pass the metaclass as metaclass not just inherite it:\nclass HostMountStateManager(metaclass\u003dHMSMMeta):\n\nalso the MHSMMeta needs to be \u0027type\u0027:\nclass HostMountStateManagerMeta(type):\n\nyour usage becomes much easier if you keep the __init__ in your class proper and just implement __call__ handling the singleton metaclass.\n\nhttps://www.codespeedy.com/create-a-singleton-in-python-using-a-metaclass/ has very clean and simple example how to do this.","accounts_in_message":[],"_revision_number":8},{"id":"1bddcf63be60557bbec4bdc072b6986c6f5c1ba3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-27 13:32:30.000000000","message":"Patch Set 8:\n\n\u003e \u003e I tried the following code with metaclass[1] with single worker\n \u003e but\n \u003e \u003e it created new manager everytime[2].\n \u003e \u003e The same thing works with my initial code producing the same\n \u003e \u003e manager.\n \u003e \u003e\n \u003e \u003e [1] http://paste.openstack.org/show/792739/\n \u003e \u003e [2] http://paste.openstack.org/show/792740/\n \u003e \n \u003e When you create the manager class, you need to pass the metaclass\n \u003e as metaclass not just inherite it:\n \u003e class HostMountStateManager(metaclass\u003dHMSMMeta):\n \u003e \n \u003e also the MHSMMeta needs to be \u0027type\u0027:\n \u003e class HostMountStateManagerMeta(type):\n \u003e \n \u003e your usage becomes much easier if you keep the __init__ in your\n \u003e class proper and just implement __call__ handling the singleton\n \u003e metaclass.\n \u003e \n \u003e https://www.codespeedy.com/create-a-singleton-in-python-using-a-metaclass/\n \u003e has very clean and simple example how to do this.\n\nHi Erno,\nIMO type does not have support for python3. If you want to use metaclass in pyhton3 then the example mentioned in the paste is correct.","accounts_in_message":[],"_revision_number":8},{"id":"ff270250c030a1cf6d2f81114cd7650e031747de","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-27 14:31:05.000000000","message":"Patch Set 8:\n\n\u003e \n \u003e Hi Erno,\n \u003e IMO type does not have support for python3. If you want to use\n \u003e metaclass in pyhton3 then the example mentioned in the paste is\n \u003e correct.\n\nwrong, that example works straight out of the box as expected at least on Python 3.7.4","accounts_in_message":[],"_revision_number":8},{"id":"cef7a3eb236dff536776bc0940abfcfebb17d78b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-27 22:46:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d9d9c20ee787979b864aa8f7ab7a5a3351fdcb60","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-27 22:47:59.000000000","message":"Patch Set 9:\n\nUpdates:\nAdded Metaclass\nAdded Tests","accounts_in_message":[],"_revision_number":9},{"id":"43f06437938b977ca7ae13e3e06b3aa62923493d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 00:07:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f912c271f4114a11b6e6ed42fe7db5e8 : FAILURE in 1h 01m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e9292aab7b64f1cb3933c7d016bae29 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35f686b42934c58ab5129b6d9c5c629 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4be738c055a4575b30ad7705fe70f79 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf134a2bb51443d8fb1ca3618798de2 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204d42040bfa413f9d1cf6e4d8f198a5 : FAILURE in 4m 09s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/ecb630053b1d49dbae24b06255cb1b69 : FAILURE in 1h 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9392615b677aadb0841996470a17df949a2a330a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-28 06:57:09.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"92a1abaacdfef01e1a85392fb204dc300426e878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:26:00.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40791f0070084107bf6cd92820c15ff9 : SUCCESS in 1h 20m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6474f246e2a74fe19329ecc8b5be6372 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9ac8246b784c55a56d430bb8ce6686 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c83e435934040148477173a37600be1 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98805749f81e4fa4b9f8d4618ac53081 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87c89b144fe47a792a266f162c04dc1 : FAILURE in 7m 25s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/c8bda37531b14b609ef52624740d28ec : SUCCESS in 1h 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"373cbf9d0a7166343f7e8dfa14b031352cb87020","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-28 12:11:27.000000000","message":"Patch Set 9: -Code-Review\n\nThanks Rajat for keeping working on this. I think at this point all but the leaky mountpoints/sockets on restart are addressed. Will spend some time for proper review today.","accounts_in_message":[],"_revision_number":9},{"id":"eb9f2dab94fb282bab61d1861dad69c7d126ef0d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-28 13:35:17.000000000","message":"Patch Set 9:\n\n\u003e Thanks Rajat for keeping working on this. I think at this point all\n \u003e but the leaky mountpoints/sockets on restart are addressed. Will\n \u003e spend some time for proper review today.\n\nThanks Erno.\nI was thinking of clearing them up during state initialization but i\u0027m still trying to figure out a way to get the mount_point_base config in fs_mount module.","accounts_in_message":[],"_revision_number":9},{"id":"52e737ec20136f5b38055798ca71079c96509e02","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-28 13:39:24.000000000","message":"Patch Set 9:\n\nI\u0027m seeing this strange failure on the docs job\nimport os_log : module not found\nin the fs_mount class.","accounts_in_message":[],"_revision_number":9},{"id":"f5224e71ed66271c6fbe507c168a6f97deeba676","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-04-28 13:46:33.000000000","message":"Patch Set 9: Code-Review-1\n\noslo.log is not specified in the glance_store requirements at all.  I think this means the logging in fs_mount.py needs to be reworked to just use \"logging\" instead like the rest of the glance_store code.","accounts_in_message":[],"_revision_number":9},{"id":"43707afea7bfae72c7cfdf32a07a551ddfc023be","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-04-28 13:52:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"99307004ae0e8ce0a210bb761c56dcff146746ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:46:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9aa9b7e74a7c4137901110f8f146a2f7 : SUCCESS in 1h 29m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df3a4531a8a94666ae7ac953cbb5480c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd5b9deb0b24059bd48660a174ef183 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bac63c62d19b4c3e82c7d3328910abff : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2352ebcd3e104815aafc057f4fbdbeeb : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd6c9c3d190462ba25e6cdbd337d8c2 : SUCCESS in 4m 36s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/61755a97110849579b1030a1143a56ba : SUCCESS in 1h 47m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"73eec3cf1fa5c542750e264ca02c851d3e849eec","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-04-29 18:19:19.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"265f20b41e87d11e0c73db7463569ed206685723","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 18:27:03.000000000","message":"Patch Set 10: Code-Review-1\n\n(7 comments)\n\nThe rootwrap filter is an issue. Other comments here are mostly just called out since other updates are needed.","accounts_in_message":[],"_revision_number":10},{"id":"80ed01f363f00f662445936e3034851b9a5792ef","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-29 19:02:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2179893eee1a2271f4f0d0a6e77934ad8377ea36","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-29 19:04:36.000000000","message":"Patch Set 11: Workflow-1\n\nThe cleanup code is ready and tested with unmounted and mounted dirs with single store as well as multi store configurations.\nThe tests still needs some working because of the cleanup code addition so W-1.\nBy the time please review the cleanup logic.","accounts_in_message":[],"_revision_number":11},{"id":"e776c1654bad6486b5779ccf5cbd2f8069d81158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:27:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f56ef2d4ed44825bb2f42c343af7a6f : SUCCESS in 1h 19m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbd26549e2cf4799b8accabf10a082a8 : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ddfbcb74f045e58b1cfe62a900b0e9 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca628dd2cb9f4cf1aa716403fb84cc08 : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48b235f6046d49f88803b44fb2e04218 : FAILURE in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c6eba493fc441784f9bd74f84be4aa : FAILURE in 4m 17s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/ecc8c66af0644d1a8f9d01067ac3bf88 : SUCCESS in 1h 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6c154c90ff83a9023a4f83eac1519f4331c94c8b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 07:05:30.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9f1682718bade38a4eeec9dce26afc4fefa5690f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 07:23:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c0160685ac04ba993d9c7bdc24bb790ddfd70d52","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 07:24:06.000000000","message":"Patch Set 13:\n\n(6 comments)\n\nHi Cyril and Sean, Thanks for the review.","accounts_in_message":[],"_revision_number":13},{"id":"fa965cf5cf87af4bbd64e9495c53479efe3e29d7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-30 07:34:00.000000000","message":"Patch Set 13: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"adb90882196216893134053026b21305441c0d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 09:13:24.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/972d13bf480f4764adba9de93cf693d3 : SUCCESS in 1h 32m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5963c22593fc405abb48050371f89f44 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76b39d1f43e4b1da0098606467d5fa7 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/070313c50b964d869d05d511f2b61237 : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ac0bbf3a1ad4d10afc046a696049ddd : FAILURE in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344410113a284120a9687853a36fdbeb : FAILURE in 4m 29s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/72ee47ec2c324671bf82bcf5180736f4 : SUCCESS in 1h 48m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5ac8c53d5345da979551f7578295c0ef58401cb1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 10:44:54.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f0c82ace93ba8867d0ca4e595755dcdf55f918ac","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 10:55:53.000000000","message":"Patch Set 14:\n\n(5 comments)\n\nThanks for the review Abhishek.","accounts_in_message":[],"_revision_number":14},{"id":"6311b166123b429c1ed9384bc5a2f8a6b59099ef","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-30 11:35:02.000000000","message":"Patch Set 13:\n\n(10 comments)\n\nFew comments inline.","accounts_in_message":[],"_revision_number":13},{"id":"5bbcefc4746f07000eb6820d6455f29b2231c042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 12:54:38.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1a293cca6044a1cb4ad58832ca85ef7 : TIMED_OUT in 2h 07m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16d265a51db94839a8c9b3da707c4991 : FAILURE in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd9c3a6d3c843e3a878837b9bb7466c : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8da4051a4ef4b1a9e6b10202ed3a47c : FAILURE in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b56ef18ff95048dfa043be3d2e4f884d : FAILURE in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397be4e0e457429a805882d16456ddd3 : SUCCESS in 4m 30s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/e721ef40b191452a829389d7db75760e : SUCCESS in 2h 05m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4ad16d1ffd1509b5a6285b68be8fdbe57b5cabfc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 13:14:48.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8c6df0c71021087815a032c8e0947f883a7cbef5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 13:16:44.000000000","message":"Patch Set 15:\n\n(8 comments)\n\nThanks for the review Erno.","accounts_in_message":[],"_revision_number":15},{"id":"f0f449038ed4c23b938145faa836981ea53147c5","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-30 13:28:55.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"7c72c55dd16485446b32dac1aad170b770f5b90b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 13:40:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a632b6ea02ddc9c44072edb413fcc14be250d2b6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 13:40:41.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c07ebf6cab9fe0176a68e821183a74903f35e314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:14:14.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1dd5137a7ef047c4803ac65dda305089 : SUCCESS in 1h 32m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/316ff648b82f4c80b9883e1913b8e65a : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a02ec0631424d699d198a80cc75bd0c : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b41b2c6edab4a4299743f8557bdbce1 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41aa63ebce264aeb9777653d86378cdc : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca5799a5e604afeb6700565f77f61a2 : SUCCESS in 4m 28s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/d0f536c5a3c9476a87f74ab7f16847df : SUCCESS in 1h 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6f18128d9942be02675b53bf406d0223de2c81d8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-30 15:14:58.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)\n\nsome nits and one error","accounts_in_message":[],"_revision_number":16},{"id":"0e702b7cc900b1538e224ed9804edf55fdc7e9d6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-30 15:24:37.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"de73b02ca365c3888c31cf672a0d97c72c7a3e37","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-30 15:28:18.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f31b8ca4dd3c89c5f70ba9293815d756f343414e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 16:57:35.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af6db41baf3046d685c61895a5e75362 : SUCCESS in 1h 26m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa1be4792ff4468c89e555b96d3fd824 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e2a1e5f9ad4cec844b3ec6e8ab5dc9 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbd7d297011245069faf4be826c82152 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7125035637864233901ed82edf62eb09 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/103bc43a19fb4badb2966271569e3f3a : SUCCESS in 6m 28s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/8dc90bbcc5bb4c858f4759a828b884a9 : SUCCESS in 1h 08m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4a2ae30b74768bce21a9358af9df637e38267815","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-30 16:59:14.000000000","message":"Patch Set 17: Code-Review+2 Review-Priority+1\n\nFew nits, but I noted down so that can be taken care while refactoring whole cinder driver during V cycle.\n\nLooks good to me and works perfectly.","accounts_in_message":[],"_revision_number":17},{"id":"a1bc1e30c3b75ea4dfec56c4779c7f3d33019035","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-05-01 11:30:31.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\nWorks for me, Thanks!","accounts_in_message":[],"_revision_number":17},{"id":"19e742e008fa560358bf2e6a11f7414d2cc2e7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 11:30:42.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"453dd19fb30680a137d9a03a2707352cd40776b9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-05-01 11:40:22.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"e0b4b131f24e7d75cbba2d012aaf09bb24f8c7d2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-01 12:45:17.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"e5e2f8f199ce261419ab0f0afd5dfe77d4dc9bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 13:32:54.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/087dd02ac55f4e7f8fb44c16fddc9b53 : SUCCESS in 1h 51m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e75d3888dedd4c26bc6a24f9f77af4e4 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a016752ff193439e846627f8c12d5252 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be8157c41b754cbbadae4663d5c7d5cb : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fee7611f52e432ab26898020ef9aa61 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b5d458f80146fba160e764b5709001 : SUCCESS in 7m 19s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/939d7b9cd22d4720a374128ec7ff5243 : FAILURE in 1h 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fe11181b9e7a15e0518ec1f7d69bdc7362c4d4d9","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-05-01 13:33:26.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\nLets try this again","accounts_in_message":[],"_revision_number":18},{"id":"1687a9b62bbd93277f0f6bef8c11882d44cc438a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 13:33:37.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"365b13e1cc1fd6429e9c0947753e34b9e9accbeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 15:31:04.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97c444cccf9343d590466cbf076d279f : SUCCESS in 1h 56m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd4b6470ea63454ab7db14a558fc64ea : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48637e130119496f9e6901fd682bf53f : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16996f4c5b5148abb8e683ed3738c0ab : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/290e1c8e0e634d4a9af0a257b50d8d06 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11493b57b9a4ff4a232b22740b01fc0 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":18},{"id":"b29d4724ac394556425ea1b39f2152794c855ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 15:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"42041ed034ee92da789f6f8eb05c8e42b2294568","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 15:32:35.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c3bf1affd2421187fdcf498ff4c930 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"f5d46996139593f6249138bcc9f09b6120f0034a","revisions":{"549397f876ba915f076e62b352e794d42c891851":{"kind":"REWORK","_number":1,"created":"2020-04-02 07:27:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/1"}}},"commit":{"parents":[{"commit":"97bb16a757ab7dd7b880c56044fbc21b17cbfbb9","subject":"Merge \"Refactor methods in cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/97bb16a757ab7dd7b880c56044fbc21b17cbfbb9"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-02 07:20:57.000000000","tz":0},"subject":"Add lock per share for  cinder nfs mount","message":"Add lock per share for  cinder nfs mount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/549397f876ba915f076e62b352e794d42c891851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/549397f876ba915f076e62b352e794d42c891851"}]},"branch":"refs/heads/master"},"6689641c411f9c6da4ac34f6e19f04cf501cae67":{"kind":"REWORK","_number":2,"created":"2020-04-02 13:51:57.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/2"}}},"commit":{"parents":[{"commit":"97bb16a757ab7dd7b880c56044fbc21b17cbfbb9","subject":"Merge \"Refactor methods in cinder store\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/97bb16a757ab7dd7b880c56044fbc21b17cbfbb9"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-02 13:51:13.000000000","tz":0},"subject":"Add lock per share for  cinder nfs mount","message":"Add lock per share for  cinder nfs mount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6689641c411f9c6da4ac34f6e19f04cf501cae67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6689641c411f9c6da4ac34f6e19f04cf501cae67"}]},"branch":"refs/heads/master"},"ab7631b31cdbe7e18883431608800a2aaef197f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-17 12:03:09.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/3"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-17 12:03:09.000000000","tz":0},"subject":"Add lock per share for  cinder nfs mount","message":"Add lock per share for  cinder nfs mount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ab7631b31cdbe7e18883431608800a2aaef197f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ab7631b31cdbe7e18883431608800a2aaef197f1"}]},"branch":"refs/heads/master"},"99e034a15f4adf1007a9a39ecccc24c430af28d4":{"kind":"REWORK","_number":4,"created":"2020-04-17 12:59:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/4"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-17 12:55:41.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/99e034a15f4adf1007a9a39ecccc24c430af28d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/99e034a15f4adf1007a9a39ecccc24c430af28d4"}]},"branch":"refs/heads/master"},"6a08a5b638e369ecb8f36c5c04e50b30677caedc":{"kind":"REWORK","_number":5,"created":"2020-04-22 10:24:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/5"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-22 10:24:41.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6a08a5b638e369ecb8f36c5c04e50b30677caedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6a08a5b638e369ecb8f36c5c04e50b30677caedc"}]},"branch":"refs/heads/master"},"ef7472d812ee10295c23d480eba3f001be9566b0":{"kind":"REWORK","_number":6,"created":"2020-04-22 13:48:35.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/6"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-22 13:48:31.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ef7472d812ee10295c23d480eba3f001be9566b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ef7472d812ee10295c23d480eba3f001be9566b0"}]},"branch":"refs/heads/master"},"631f6d27f373bcbe871a78afea98047faa1ddcf4":{"kind":"REWORK","_number":7,"created":"2020-04-22 13:55:25.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/7"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-22 13:55:21.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/631f6d27f373bcbe871a78afea98047faa1ddcf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/631f6d27f373bcbe871a78afea98047faa1ddcf4"}]},"branch":"refs/heads/master"},"fd4b9a01382da15da5bb5bbf9c3dd7137507b777":{"kind":"REWORK","_number":8,"created":"2020-04-24 09:00:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/8"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-24 09:00:02.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fd4b9a01382da15da5bb5bbf9c3dd7137507b777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/fd4b9a01382da15da5bb5bbf9c3dd7137507b777"}]},"branch":"refs/heads/master"},"79fede0ba6b2f61b1a8983c8a4f0c80a1fbc798d":{"kind":"REWORK","_number":9,"created":"2020-04-27 22:46:04.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/9"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-27 22:41:17.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/79fede0ba6b2f61b1a8983c8a4f0c80a1fbc798d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/79fede0ba6b2f61b1a8983c8a4f0c80a1fbc798d"}]},"branch":"refs/heads/master"},"87966ce11fcb4475e790be117f052d5509518bb8":{"kind":"REWORK","_number":10,"created":"2020-04-28 13:52:10.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/74/716874/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/10"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2020-04-28 13:51:36.000000000","tz":-240},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/87966ce11fcb4475e790be117f052d5509518bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/87966ce11fcb4475e790be117f052d5509518bb8"}]},"branch":"refs/heads/master"},"393106e9d7181719f902e2d4212dafc5b15912e3":{"kind":"REWORK","_number":11,"created":"2020-04-29 19:02:01.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/11"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-29 18:56:53.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/393106e9d7181719f902e2d4212dafc5b15912e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/393106e9d7181719f902e2d4212dafc5b15912e3"}]},"branch":"refs/heads/master"},"2f85bdaa1aca5575f09eb826230b1dfeb10d780d":{"kind":"REWORK","_number":12,"created":"2020-04-30 07:05:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/12"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-30 07:05:07.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2f85bdaa1aca5575f09eb826230b1dfeb10d780d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2f85bdaa1aca5575f09eb826230b1dfeb10d780d"}]},"branch":"refs/heads/master"},"195b9c34091d6545119de18998a8c7b683190b4c":{"kind":"REWORK","_number":13,"created":"2020-04-30 07:23:27.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/13"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-30 07:23:05.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/195b9c34091d6545119de18998a8c7b683190b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/195b9c34091d6545119de18998a8c7b683190b4c"}]},"branch":"refs/heads/master"},"91465a38dfbb5bb7a2be038242d22094fd00ac7c":{"kind":"REWORK","_number":14,"created":"2020-04-30 10:44:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/14"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-30 10:44:33.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/91465a38dfbb5bb7a2be038242d22094fd00ac7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/91465a38dfbb5bb7a2be038242d22094fd00ac7c"}]},"branch":"refs/heads/master"},"d6176c2bedd77f37972bbd45eb1b3f1743083272":{"kind":"REWORK","_number":15,"created":"2020-04-30 13:14:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/15"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-30 13:13:58.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d6176c2bedd77f37972bbd45eb1b3f1743083272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d6176c2bedd77f37972bbd45eb1b3f1743083272"}]},"branch":"refs/heads/master"},"3e33cfad394afc6a6b5183fac45e7919c8a95d52":{"kind":"REWORK","_number":16,"created":"2020-04-30 13:40:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/16"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-30 13:39:49.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3e33cfad394afc6a6b5183fac45e7919c8a95d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3e33cfad394afc6a6b5183fac45e7919c8a95d52"}]},"branch":"refs/heads/master"},"d94f57c0e95648cc9034b2e69097d9573165ad12":{"kind":"REWORK","_number":17,"created":"2020-04-30 15:24:37.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/17"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-04-30 15:19:23.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d94f57c0e95648cc9034b2e69097d9573165ad12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d94f57c0e95648cc9034b2e69097d9573165ad12"}]},"branch":"refs/heads/master"},"f5d46996139593f6249138bcc9f09b6120f0034a":{"kind":"REWORK","_number":18,"created":"2020-05-01 11:40:22.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/716874/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/74/716874/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/74/716874/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/74/716874/18"}}},"commit":{"parents":[{"commit":"2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2aa892d4dfb1cf8beb5f9bdba4e468486d4bae44"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-05-01 11:35:01.000000000","tz":0},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f5d46996139593f6249138bcc9f09b6120f0034a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f5d46996139593f6249138bcc9f09b6120f0034a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
