)]}'
{"id":"openstack%2Fglance_store~726648","triplet_id":"openstack%2Fglance_store~stable%2Ftrain~I10b5ea4943ad9388ce1962eae9204b905c8ff24a","project":"openstack/glance_store","branch":"stable/train","topic":"bug/1870289","hashtags":[],"change_id":"I10b5ea4943ad9388ce1962eae9204b905c8ff24a","subject":"Add config for cinder mounting needs","status":"MERGED","created":"2020-05-10 19:43:00.000000000","updated":"2020-08-25 15:42:37.000000000","submitted":"2020-08-25 15:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726648-1598370007298-1545f216","meta_rev_id":"e733e8c6623dd2099515637033d3606c897561fa","_number":726648,"virtual_id_number":726648,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 15:40:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2020-08-25 13:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-06 20:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-06 20:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2020-08-25 13:51:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 05:22:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-20 07:14:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-08-25 13:51:25.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-25 15:40:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77051c135cb63efff26185852423c9e41f6f81a9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-05-10 19:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bb5c2dbac09124b9a45bda625e3ad405ae5bb1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 21:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/74291ed3469543c0af4626a5ffe85ee0 : SUCCESS in 1h 38m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/028ec9d21348467fb8ee882ffa7c8396 : SUCCESS in 1h 21m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca4c23edcdab424f828da4cda4e5ebb4 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4bddf6a47c4f3f9c2ed82a5df3f85e : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c16436d7262c4a5baa931df179444e40 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01963b3f35f949358578dda96631626c : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/158f34b6565b4ad69fcc8086cab741b4 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb93f5503224e33a3036b91179a498d : SUCCESS in 4m 01s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/f8c56a7a9d1d4efd93bebc663b492d2b : SUCCESS in 1h 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68df62646bc613ab5ae9071475e3a51962a00d92","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-18 05:22:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf0463feffe2560f414ee20d4290a2479a5a3c74","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-05-20 07:14:37.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"42703e18896a5c919872bb6367e151f60a9b0334","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-07-06 20:57:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6361f4984ec31373eda972de704fdf4e78aebae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 22:39:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a1c1a90c68124dad8afa05150ac979e6 : SUCCESS in 1h 23m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4ea7e43138a4936891df1caf5eef786 : SUCCESS in 1h 38m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2678fa1e51ea47ab9bdbe3c066fc3d12 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3381f9fc1624d568c4dccdf9abd9124 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52d3c41b24854f0c93bc302255930068 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958fc5601fad4444b677068f586228d2 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bad6e433d3b47a3819ac397c2112d17 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35a9731c3da44f6b2db8f33ac452a3c : SUCCESS in 3m 57s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/af6afb3165f24e289d5059970f793f17 : FAILURE in 1h 04m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e01d4847f2f2fbfe87a4fdc734af6ae6ee050bc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-25 13:51:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b20681d25005247afeec42c96356a901cd0abf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:51:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab7371c934ce00b2434d266284b06e80ef656520","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:40:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7ad913d29cb10f4f7f0325525927ab27857b074c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:40:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bdc3bd6d61ce4db68fae413d830aa073 : SUCCESS in 1h 41m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6855e826949442d28998b747b16bad47 : SUCCESS in 1h 24m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ad091c9edcd4a5794d5729646cc4788 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee035c02d074969bb8bc4c0a9f84f23 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86d8dd91d1064fd49657ca604aafe61a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec50a9a8fcd746edb79b437aa9748761 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88fbd0e5a541468fbad53985ab9cca52 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a43e445cad4b9da941535b7a369d10 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"93ac4dc461d71b6de000bec4862a2331e78d3960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:42:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2298b5b18a428c839f1ba6dae2eb8f : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6b7be68f8bfd76b107ff9888ea08bd9441f03a4","revisions":{"9cd392a2f2150f0c9e22bfc377fa8fa7188f207a":{"kind":"REWORK","_number":1,"created":"2020-05-10 19:43:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/48/726648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/48/726648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/726648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/726648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/726648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/48/726648/1"}}},"commit":{"parents":[{"commit":"d052ce57b5cfc66a4b3b8030a14db3d46c312990","subject":"Image upload fails if cinder multipath is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d052ce57b5cfc66a4b3b8030a14db3d46c312990"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-02-22 16:30:26.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2020-05-10 19:41:47.000000000","tz":120},"subject":"Add config for cinder mounting needs","message":"Add config for cinder mounting needs\n\nThis patch adds a config option \u0027mount_point_base\u0027 by which we can\nprovide the mount point for *fs volumes.\nThe different types of FS drivers will have separate directory for\nmounting their volumes,\nEg: \u003cmount_point_base\u003e/nfs, \u003cmount_point_base\u003e/xyzfs etc.\nThe default mount point for *fs volumes is /var/lib/glance/mnt/*fs.\n\nCloses-Bug: #1866966\n\nChange-Id: I10b5ea4943ad9388ce1962eae9204b905c8ff24a\n(cherry picked from commit 1a656163c9d68520ab26035d350f2568cbfb881f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9cd392a2f2150f0c9e22bfc377fa8fa7188f207a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9cd392a2f2150f0c9e22bfc377fa8fa7188f207a"}]},"branch":"refs/heads/stable/train"},"c6b7be68f8bfd76b107ff9888ea08bd9441f03a4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-06 20:57:32.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/48/726648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/48/726648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/726648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/726648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/48/726648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/48/726648/2"}}},"commit":{"parents":[{"commit":"a6227665ef8ec654ca342362996b1cc40357a290","subject":"Fix: API returns 503 if one of the store is mis-configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a6227665ef8ec654ca342362996b1cc40357a290"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-02-22 16:30:26.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2020-07-06 20:43:53.000000000","tz":120},"subject":"Add config for cinder mounting needs","message":"Add config for cinder mounting needs\n\nThis patch adds a config option \u0027mount_point_base\u0027 by which we can\nprovide the mount point for *fs volumes.\nThe different types of FS drivers will have separate directory for\nmounting their volumes,\nEg: \u003cmount_point_base\u003e/nfs, \u003cmount_point_base\u003e/xyzfs etc.\nThe default mount point for *fs volumes is /var/lib/glance/mnt/*fs.\n\nCloses-Bug: #1866966\n\nChange-Id: I10b5ea4943ad9388ce1962eae9204b905c8ff24a\n(cherry picked from commit 1a656163c9d68520ab26035d350f2568cbfb881f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c6b7be68f8bfd76b107ff9888ea08bd9441f03a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c6b7be68f8bfd76b107ff9888ea08bd9441f03a4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
