)]}'
{"id":"openstack%2Fglance_store~726914","triplet_id":"openstack%2Fglance_store~stable%2Fussuri~I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6","project":"openstack/glance_store","branch":"stable/ussuri","topic":"bug/1870289","hashtags":[],"change_id":"I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6","subject":"Add lock per share for cinder nfs mount/umount","status":"MERGED","created":"2020-05-11 17:44:05.000000000","updated":"2020-07-02 21:37:03.000000000","submitted":"2020-07-02 21:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726914-1593725733938-e155475f","meta_rev_id":"63ce2f5ce69aabecab8cdece379959bc1f369c21","_number":726914,"virtual_id_number":726914,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 21:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-02 20:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-18 05:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-20 07:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-02 20:10:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 05:22:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-20 07:17:21.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-07-02 20:10:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-02 21:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1f29949862b0abcad566a12f8dc745321232743","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-05-11 17:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ede5b2784b9f28afe389743d972710d465c2e01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 19:30:07.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/610e576bc60d44ed8bbf0f5751b38d5d : SUCCESS in 1h 42m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c91bb5a732c464b81c19666873d8ed7 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816c190dff0d426dbf6b112b6a7ce11e : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/242f626378f24fbe834b37305e2e0c6a : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfe8b33beb9e48949f0087b621844ccd : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2b8e259ede4adaa344c9a2c979e869 : SUCCESS in 4m 07s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/cf36d3532abd4c5cbc8e7a16f2f142ad : FAILURE in 1h 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b50289bc9261d6684b75395a2e4333a9716a39ce","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-05-11 20:42:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91da57f91c4de51a54644922bc603339c5c2e245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:29:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43cdd88a5a8e4823b0091f8c03feb178 : SUCCESS in 1h 40m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecd69945c5c64f13a395270ec9671815 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff573a543f34d78a9234c3a274e062b : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b60a4f18964767a99887e6635abdfe : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d14620c7298648b6915fe159e2bbcde6 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa7c52e6cf34338bc79e1ab18dd612f : SUCCESS in 3m 45s\n- legacy-tempest-dsvm-full-ceph-plugin-src-glance_store https://zuul.opendev.org/t/openstack/build/1924cd37ac414be29a027cf4ba1592b4 : FAILURE in 1h 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee86cd5633ac48e520268dc9813be7bd87668cd3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-18 05:22:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f5b08c79077d00a13c739478e1e5d5ff522657d3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-05-20 07:17:21.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"8dee306b99c1815c50ccc717bde12cb50213d269","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-02 20:10:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"306de969aee5e037076f7ab77c34ded31a249611","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:10:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52fc32d799f3cee7d7c362c337005535aea77c74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 21:35:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbe54226727742e685061089a4283290 : SUCCESS in 1h 19m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f9cc62147e4a77b9a55b1f0eccdbdd : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1f739ed494466eb708381fbcd983aa : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac7f0131fe774666ad918cc1fbade8c6 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d55abe50b89649d7b0820efb5fb82758 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da97d932fefc4c84918a73a5b98c4202 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"cb42132b5274a5e5dba79281c286f85396db0774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 21:35:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6cb5368040fb9ad0b75e7a71f2516f20675fd2cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 21:37:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ba50c2820242619eec0f60f04b7034 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b37e7f2f3c3faf3b6e4d11eeb3e866211e13b85e","revisions":{"44912c039435f1dca7db17fae74e825126a82d03":{"kind":"REWORK","_number":1,"created":"2020-05-11 17:44:05.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/14/726914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/726914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/726914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/726914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/726914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/726914/1"}}},"commit":{"parents":[{"commit":"17b66ab6a50f8dda28fc2f4c02489bb5ba4241e3","subject":"Fix: API returns 503 if one of the store is mis-configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/17b66ab6a50f8dda28fc2f4c02489bb5ba4241e3"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2020-05-11 17:43:37.000000000","tz":120},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n(cherry picked from commit f5d46996139593f6249138bcc9f09b6120f0034a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/44912c039435f1dca7db17fae74e825126a82d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/44912c039435f1dca7db17fae74e825126a82d03"}]},"branch":"refs/heads/stable/ussuri"},"b37e7f2f3c3faf3b6e4d11eeb3e866211e13b85e":{"kind":"REWORK","_number":2,"created":"2020-05-11 20:42:07.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/14/726914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/14/726914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/726914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/726914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/14/726914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/14/726914/2"}}},"commit":{"parents":[{"commit":"17b66ab6a50f8dda28fc2f4c02489bb5ba4241e3","subject":"Fix: API returns 503 if one of the store is mis-configured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/17b66ab6a50f8dda28fc2f4c02489bb5ba4241e3"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-01 14:18:44.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2020-05-11 20:41:58.000000000","tz":120},"subject":"Add lock per share for cinder nfs mount/umount","message":"Add lock per share for cinder nfs mount/umount\n\nWith cinder configured as glance store and nfs as cinder backend,\nif we create multiple images concurrently into the same share,\nthere might be race conditions.\nThis patch ensures a lock per export to avoid the race conditions.\n\nThis patch also introduces a reference counting mechamism for unmounting\nwhich will unmount the share if no active thread/process is using it.\n\nCloses-Bug: #1870289\n\nChange-Id: I9197f64e29a0ae2e0a58186f1a70aa134f7f1db6\n(cherry picked from commit f5d46996139593f6249138bcc9f09b6120f0034a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b37e7f2f3c3faf3b6e4d11eeb3e866211e13b85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b37e7f2f3c3faf3b6e4d11eeb3e866211e13b85e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
