)]}'
{"id":"openstack%2Fglance_store~746556","triplet_id":"openstack%2Fglance_store~master~I0ba36d1095886fd1d96926af37a3a65a178c004f","project":"openstack/glance_store","branch":"master","topic":"746556","hashtags":[],"change_id":"I0ba36d1095886fd1d96926af37a3a65a178c004f","subject":"Support Cinder multiple stores","status":"MERGED","created":"2020-08-17 16:03:14.000000000","updated":"2020-08-25 15:09:13.000000000","submitted":"2020-08-25 15:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746556-1598367990319-9b0f7ed9","meta_rev_id":"f1e2ae321b35ee3c6575234ab8dcf4f74f32e75c","_number":746556,"virtual_id_number":746556,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 15:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-24 21:28:28.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-08-25 13:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-25 05:46:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-08-25 13:48:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-25 05:46:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 10:13:25.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-08-20 10:13:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-08-25 05:46:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-25 13:48:23.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-25 15:06:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86690eb12d5b9a10a4cdfcc7825c92d0ca8bb162","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-17 16:03:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cec5574ed973558325d7caed3172e8964d068ba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-17 16:44:21.000000000","message":"Patch Set 1: Code-Review-1 Review-Priority+1\n\n(7 comments)\n\nThank you for working on this, some suggestions noted inline.","accounts_in_message":[],"_revision_number":1},{"id":"4a38feb5d9d9fd11e981e0f18f2f22c1e2436bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 18:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bb69a17560746809639e6256814693f : SUCCESS in 1h 16m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae3e4d9c26c4fe39677e66b390e30ad : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d9a83b2e934502b518312a5e90b40c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/920342b4a2f84b6ea25030dabff1c52c : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/838e9050449f4738989619ff9f11c58d : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98bfca6a16043e4b46cb9fb71454cfc : SUCCESS in 4m 24s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b993492352b4081abb8768aa26b71b2 : SUCCESS in 1h 48m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d51fbe192e5ff0d01225fc5cad99136d52a22426","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 01:30:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d03325a574c77b928b85bb0ea91c89038e9c248","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 01:30:53.000000000","message":"Patch Set 2:\n\n(7 comments)\n\nThanks Abhishek for the review.","accounts_in_message":[],"_revision_number":2},{"id":"0cb7a1504ac03187f8679b852cfae115de58473b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 02:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9b3d4872136410d886787b324614a5d : SUCCESS in 1h 06m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e55163d0fb644e9ab39a8c2cf22e1b5 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a32c4fba284d70afbbf3554f161f08 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a285bd1d74543e1a03f24ccd0183f90 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54493ea743ac486a9524d5d80901802e : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e352651d1bf642ec8df9c6f615f9e961 : SUCCESS in 3m 20s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/63eea0085dfc4c02a106fad1b7f67a32 : SUCCESS in 1h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"944c54c61f8cefc2d9d91e4d222a02d952349205","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 04:58:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3387c60a1d69caa81d40d803f2e118e728aa842b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 06:33:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b17fffb1ff89409ebadb54f60167e426 : SUCCESS in 1h 24m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acfa764576674ff6a3d4ee018e63cf2c : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8ef12f903a4d9297c8cdf7fc2d30a4 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/290ea110a0fa477e951aaf7bfd0183b6 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e201c0fa1ba42119b3a43f279ffc332 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbaa1dc549f54b73bd07023ff4e67e75 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8450adb723cf4257998779ef4a35e892 : SUCCESS in 4m 04s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/8c337b48c4c34dba8412a5e748c7657f : SUCCESS in 1h 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb08f0f6b28de3f5c7734f1f35c60a796c83d583","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-20 13:47:11.000000000","message":"Patch Set 3: Code-Review+1 Review-Priority+1\n\n(1 comment)\n\nCode looks good to me, just one comment inside.","accounts_in_message":[],"_revision_number":3},{"id":"e5314cbadd9b1dafd2362beabea35535b10164be","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-24 17:35:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nQuick change suggestion inline and we definitely need to change the logging level. Other than that looks pretty good.","accounts_in_message":[],"_revision_number":3},{"id":"49ddd2e646368c5ad7b95dd9f326f5bd7e2d8a09","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-24 17:40:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8fa700e68549830cd557aff654c2ff473f2d06d0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-24 17:49:18.000000000","message":"Patch Set 3: -Code-Review\n\ngood suggestions Erno!!","accounts_in_message":[],"_revision_number":3},{"id":"9ff90856b8b35d6e55ab98ea2aeb33df56b9eea1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 19:34:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cc794cee4e695f8aa9b2a101837635e39da59c15","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 19:35:12.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nThanks Abhishek and Erno","accounts_in_message":[],"_revision_number":4},{"id":"a6b9630bdf0d77eedbcbf2db92811958db13fedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:14:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd07b5e3e11c41dc8f4ab39252eea493 : SUCCESS in 53m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edd598a3d97347d5b7bd330dd1b7e2ef : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48e82a4b2214ceb91ddddee06ba75e1 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45b15f64aa5d485c9f9f702581747952 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/971f186ad1e043b5abeb3f0ac53618bb : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996c5f09a1a44e3dac45aa792c5264f9 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2850abc3f204438292a939fa350bc7ef : SUCCESS in 5m 24s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/60042a5c07d54384b2140af945c7a50a : SUCCESS in 1h 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ae4917e28fc317d7b5c02a126ba0dd6c36fd4a08","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-24 21:26:14.000000000","message":"Patch Set 4: Code-Review-1 Review-Priority+1\n\nLast reno is added by mistake","accounts_in_message":[],"_revision_number":4},{"id":"9c2833e1f5ca7316b4bcf97ef50d132c808b2313","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 21:27:59.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ff8168049a911a329ac1696c67a0d234005c2d4f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-24 21:28:28.000000000","message":"Patch Set 5:\n\n\u003e Last reno is added by mistake\n\nMy bad, removed.","accounts_in_message":[],"_revision_number":5},{"id":"f3570500598acadc373e73a18ea4446866fc6a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 23:36:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/accc6804401041b38ce95a38fd1bb72c : SUCCESS in 1h 36m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41187b753c714adcbf27022e06879919 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473ac242f69f4b4497c34a95bb734aa6 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09dcfa30a64140d88628f723c2888232 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0025b3e8cb054e02841ba532e0322796 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bbf340f36ed4481b51a9fa5d4513003 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb0ad09029a4aeda69fcee921fce734 : SUCCESS in 3m 23s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/c4ffbbef658e464f889e3bef0af13e32 : TIMED_OUT in 2h 05m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"343755a0a6d4759566cbfc0b2afb9047609583dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-25 05:46:56.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1\n\nThank you Rajat!!!","accounts_in_message":[],"_revision_number":5},{"id":"e04171e331f9e27605e2cc44f3a9188ec5105d78","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-25 13:48:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks Rajat","accounts_in_message":[],"_revision_number":5},{"id":"e0706ccd582fb5b0a1e2f2268296e23cbe0f2595","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:48:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0b80158399f12862fc588c1dedc18a7dfa67db93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:06:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5eca3e4ae4948b09535ece733b67800 : SUCCESS in 1h 14m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec5e0bca706244a2b00fac24b917f2f1 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937af3af909f4029afae2d1cf640d5d4 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b11c59766b63427ca366e4d65da3a6fb : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b92940bb9c84b2d98bfac3ad4d1069f : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c31c19006c4602bdbcb57199afdd21 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fdb0179d3634eaab9f75454e58cf4c8 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":5},{"id":"2010779e07c5ecd0e0621d23e51d4bdd617d87a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:06:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"148de2c3b36869279492d97a5582babd0f60ab4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:09:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7032b021cce04d88847ca57dd7adcc32 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee0bc6cc487a4512a8662cd8bb14c902 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"eec0948343d1430515e7ef5e90ec9063044c17a2","revisions":{"105c703cf1ba9f0226bde4b039e441e210e2cefc":{"kind":"REWORK","_number":1,"created":"2020-08-17 16:03:14.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/746556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/56/746556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/56/746556/1"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-17 15:59:29.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-17 15:59:56.000000000","tz":0},"subject":"Support Cinder multiple stores","message":"Support Cinder multiple stores\n\nThis patch adds support for configuring multiple cinder stores.\nThe following changes are part of this patch:\n1) location URL format: previously cinder://\u003cvolume-id\u003e, the format is\nnow changed to cinder://\u003cstore-name\u003e/\u003cvolume-id\u003e\n2) Check for configured volume types: During initialization of service,\nall volume types configured in the cinder store(s) will be checked if\nthey exist in deployment or not, the store with the wrong type will be\ndisabled.\n3) Update legacy images: This change is partially depedent on the glance\nchanges[1] which trigger updating the URL for legacy images and the\ncinder store checks the configured stores volume type against the existing\nimage-volume\u0027s type and update it respectively.\n\n[1] https://review.opendev.org/#/c/742373/\n\nChange-Id: I0ba36d1095886fd1d96926af37a3a65a178c004f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/105c703cf1ba9f0226bde4b039e441e210e2cefc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/105c703cf1ba9f0226bde4b039e441e210e2cefc"}]},"branch":"refs/heads/master"},"caab18fe334a92eb9abdb3a0fc5d83d1ab4a6bcd":{"kind":"REWORK","_number":2,"created":"2020-08-19 01:30:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/746556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/56/746556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/56/746556/2"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-17 15:59:29.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-19 01:25:26.000000000","tz":0},"subject":"Support Cinder multiple stores","message":"Support Cinder multiple stores\n\nThis patch adds support for configuring multiple cinder stores.\nThe following changes are part of this patch:\n1) location URL format: previously cinder://\u003cvolume-id\u003e, the format is\nnow changed to cinder://\u003cstore-name\u003e/\u003cvolume-id\u003e\n2) Check for configured volume types: During initialization of service,\nall volume types configured in the cinder store(s) will be checked if\nthey exist in deployment or not, the store with the wrong type will be\ndisabled. if cinder store is configured using traditional/old way\n(i.e. using \u0027stores\u0027 config option) then this validation will not be performed.\n3) Update legacy images: This change is partially depedent on the glance\nchanges[1] which trigger updating the URL for legacy images and the\ncinder store checks the configured stores volume type against the existing\nimage-volume\u0027s type and update it respectively.\n\n[1] https://review.opendev.org/#/c/742373/\n\nImplements: blueprint multiple-cinder-backend-support\nChange-Id: I0ba36d1095886fd1d96926af37a3a65a178c004f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/caab18fe334a92eb9abdb3a0fc5d83d1ab4a6bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/caab18fe334a92eb9abdb3a0fc5d83d1ab4a6bcd"}]},"branch":"refs/heads/master"},"b955bc9e83f552d1b7b5a489760a1671d41db2a0":{"kind":"REWORK","_number":3,"created":"2020-08-19 04:58:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/746556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/56/746556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/56/746556/3"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-17 15:59:29.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-19 04:57:44.000000000","tz":0},"subject":"Support Cinder multiple stores","message":"Support Cinder multiple stores\n\nThis patch adds support for configuring multiple cinder stores.\nThe following changes are part of this patch:\n1) location URL format: previously cinder://\u003cvolume-id\u003e, the format is\nnow changed to cinder://\u003cstore-name\u003e/\u003cvolume-id\u003e\n2) Check for configured volume types: During initialization of service,\nall volume types configured in the cinder store(s) will be checked if\nthey exist in deployment or not, the store with the wrong type will be\ndisabled. if cinder store is configured using traditional/old way\n(i.e. using \u0027stores\u0027 config option) then this validation will not be performed.\n3) Update legacy images: This change is partially depedent on the glance\nchanges[1] which trigger updating the URL for legacy images and the\ncinder store checks the configured stores volume type against the existing\nimage-volume\u0027s type and update it respectively.\n\n[1] https://review.opendev.org/#/c/742373/\n\nImplements: blueprint multiple-cinder-backend-support\nChange-Id: I0ba36d1095886fd1d96926af37a3a65a178c004f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b955bc9e83f552d1b7b5a489760a1671d41db2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/b955bc9e83f552d1b7b5a489760a1671d41db2a0"}]},"branch":"refs/heads/master"},"ca2929181ac0b940342493cde928ea61f8dc982f":{"kind":"REWORK","_number":4,"created":"2020-08-24 19:34:11.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/746556/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/56/746556/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/56/746556/4"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-17 15:59:29.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-24 19:29:51.000000000","tz":0},"subject":"Support Cinder multiple stores","message":"Support Cinder multiple stores\n\nThis patch adds support for configuring multiple cinder stores.\nThe following changes are part of this patch:\n1) location URL format: previously cinder://\u003cvolume-id\u003e, the format is\nnow changed to cinder://\u003cstore-name\u003e/\u003cvolume-id\u003e\n2) Check for configured volume types: During initialization of service,\nall volume types configured in the cinder store(s) will be checked if\nthey exist in deployment or not, the store with the wrong type will be\ndisabled. if cinder store is configured using traditional/old way\n(i.e. using \u0027stores\u0027 config option) then this validation will not be performed.\n3) Update legacy images: This change is partially depedent on the glance\nchanges[1] which trigger updating the URL for legacy images and the\ncinder store checks the configured stores volume type against the existing\nimage-volume\u0027s type and update it respectively.\n\n[1] https://review.opendev.org/#/c/742373/\n\nImplements: blueprint multiple-cinder-backend-support\nChange-Id: I0ba36d1095886fd1d96926af37a3a65a178c004f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ca2929181ac0b940342493cde928ea61f8dc982f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ca2929181ac0b940342493cde928ea61f8dc982f"}]},"branch":"refs/heads/master"},"eec0948343d1430515e7ef5e90ec9063044c17a2":{"kind":"REWORK","_number":5,"created":"2020-08-24 21:27:59.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/746556/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/56/746556/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/56/746556/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/56/746556/5"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-08-17 15:59:29.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-08-24 21:27:51.000000000","tz":0},"subject":"Support Cinder multiple stores","message":"Support Cinder multiple stores\n\nThis patch adds support for configuring multiple cinder stores.\nThe following changes are part of this patch:\n1) location URL format: previously cinder://\u003cvolume-id\u003e, the format is\nnow changed to cinder://\u003cstore-name\u003e/\u003cvolume-id\u003e\n2) Check for configured volume types: During initialization of service,\nall volume types configured in the cinder store(s) will be checked if\nthey exist in deployment or not, the store with the wrong type will be\ndisabled. if cinder store is configured using traditional/old way\n(i.e. using \u0027stores\u0027 config option) then this validation will not be performed.\n3) Update legacy images: This change is partially depedent on the glance\nchanges[1] which trigger updating the URL for legacy images and the\ncinder store checks the configured stores volume type against the existing\nimage-volume\u0027s type and update it respectively.\n\n[1] https://review.opendev.org/#/c/742373/\n\nImplements: blueprint multiple-cinder-backend-support\nChange-Id: I0ba36d1095886fd1d96926af37a3a65a178c004f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/eec0948343d1430515e7ef5e90ec9063044c17a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/eec0948343d1430515e7ef5e90ec9063044c17a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
