)]}'
{"id":"openstack%2Fglance_store~746579","triplet_id":"openstack%2Fglance_store~master~I7f0bffda222b663d4316c5d6c03fdbd0d3337035","project":"openstack/glance_store","branch":"master","topic":"rbd_resize","hashtags":[],"change_id":"I7f0bffda222b663d4316c5d6c03fdbd0d3337035","subject":"Ramp up rbd resize to avoid excessive calls","status":"MERGED","created":"2020-08-17 18:01:24.000000000","updated":"2020-11-12 10:52:49.000000000","submitted":"2020-08-21 09:08:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746579-1598000925186-bd91aa04","meta_rev_id":"48f49f9f0734f0743198315bb9f80bc292c2167a","_number":746579,"virtual_id_number":746579,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 09:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2020-08-21 07:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2020-08-20 15:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2020-08-21 07:54:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2020-08-21 05:51:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 14:40:40.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-08-20 15:05:14.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-08-21 07:54:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-21 09:08:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c02288862e5b3a4450fc527ea20eba0c85aec559","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-17 18:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c276a899ee0e5a9194781c7d732aa91f496a1cbc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-17 18:40:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"780ef6d6aa9640467677e5904e4b1e8eb46f8560","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-17 19:47:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nSome thoughts inline and a couple places I think the test assertions can be a lot more stringent to help ensure the intent is being realized.","accounts_in_message":[],"_revision_number":2},{"id":"99a6619444aa93ad51af13588c5a5e7f80d5b401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 20:13:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b9aa4f33d9b4e6a86e9337ee5cb4b8c : SUCCESS in 1h 24m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dacf182079646eb8f95290b8e8627da : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58091b480d664aba95cb646d370e05f9 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82dc270103c24d22a85e2118cdc95701 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79880f8e58214290b550f8dcffa02bdf : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4843dea37f449698e39f11e71a6967 : SUCCESS in 3m 35s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4652838fade74dae8d68db1737da5ff0 : SUCCESS in 1h 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5934e13fa163518ea37cf40c17d933fc220fd6e9","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-18 12:58:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8def5b4fb7b2211a2f19de8a9af5013622709da","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-18 14:43:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4f146a9cfeea4bc0f1d2509cb252b5c00071d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:44:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09749748176545c0b1133b34ac809e5b : SUCCESS in 1h 31m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/113f9245f972487eacf2247811b67cc4 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8004f273b0491e8cb2ddadb5890e88 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/becdf7d5f73f42fe856c3a7080ded688 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71900d87c0c4001bf7fa88b6569b2a3 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b9aa06568848e5afe05c5333bc8025 : SUCCESS in 6m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/96899f2f56a947a6b474d97946380c00 : SUCCESS in 1h 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"83d9780b121141569705a801c0785284f571efd2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 14:50:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fdf5e974bac804ed892a70868789005187bd0ff7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-18 16:02:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0ccab59bcd5aee8e68b90c107499f9f6bd5eb150","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-18 16:05:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"923b18c39fc9befac35dfaa57df49f3eff68d6ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 17:45:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f4914a2158d46b987095d63667a1f4f : SUCCESS in 1h 18m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e5706c9e10b4588ac1541728005c488 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd2d56c99d449ea80f604248b58bb8e : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d6a7f55cbe485696ea7981bce67c7c : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/025c3bf5faa747c79a35477574de0996 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d5563655304430b74e824995b12b4a : SUCCESS in 4m 26s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e4b544015044947a4f36f03693da90c : SUCCESS in 1h 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c7d4666d5856a5ad528f560ab7098ef22e48f972","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-20 15:05:14.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks good to me. You\u0027re not verifying the 8G ceiling, but now that it\u0027s just a min() call it\u0027s safer to assume that it works, IMHO. I\u0027ll bring the FakeData generator over to glance_store to refactor this and will add a test case for the ceiling in the process.","accounts_in_message":[],"_revision_number":5},{"id":"250c4c65d97e474b306c0837fb9fde83804e57dc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-20 15:45:58.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"1b6a0a9e04dae462bebca95512b6951bf4c1fbd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 17:07:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6532d52a343c4071b526635118271039 : RETRY_LIMIT in 1m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe1f948215764d32a93672eebcc5aeb1 : RETRY_LIMIT in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b063cd14ae495cba2bb3328e4e8c06 : RETRY_LIMIT in 2m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4256ab48374e868047bcd0a205e61c : RETRY_LIMIT in 1m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7b9fa028074e6886c1f485d74dd020 : RETRY_LIMIT in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c4223a66a342dc87006c948cae41a6 : RETRY_LIMIT in 1m 53s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f55486940a4432492cfd2fca32f389b : RETRY_LIMIT in 1m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4b3165cf2ffd4487e7806d81680cf451851f44f7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-21 05:45:07.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8c829603d4e90e6f254926e08024ebbf2422c138","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-21 05:51:15.000000000","message":"Patch Set 6: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e0d050da9fb4e02a3ad777ec08839000c5c110cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 07:26:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1aee9dfacad344dda961310be396b085 : SUCCESS in 1h 08m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8123d2afe6c44e938442405c97af3394 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8822ee362c4d9e9c9074488daa8fb4 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7d60d8e74f34de89b24d2da8c83d104 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/361b20e0eedd46c7840eaf704aa4823c : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de24cff6c8524b3abedcc3adfdc6bc38 : SUCCESS in 3m 20s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/cfab1eb4497141e4a5bf5e99f68c41ab : SUCCESS in 1h 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f904518e934c4224cad20c8878c553700416c21e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-21 07:54:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"11ea1356d89fde8d1d740578c4e6d619d7c24ecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 07:55:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cd4d502dbff0373785ba0c123ad25fb0245a1c3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:08:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94cd9deec9fe461b966eeafc42e30260 : SUCCESS in 1h 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8e93dd3735a41fd9328960b70e18732 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa9196425884500b2e528db8b488e9b : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b28d9a482a5b4ada808777fbf312c677 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/418bc1f7d00342d8beb917023d0c6300 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd7d9728c8442b0ab2ba8db215497ab : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":6},{"id":"3dd80f91be6db7ef63f3a40377e846eae74a911f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:08:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"8388702c61f6acb649aa95b4d0487350c9842e8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:11:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd1a7ba58524dfcb328c1398266fbda : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":6},{"id":"dca55dd96d5cd624e8cb4ee4d3b832f8ad26c46f","author":{"_account_id":26498,"name":"liumengke","email":"liumk@rc.inesa.com","username":"liumk"},"date":"2020-08-25 03:02:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"825c539a0eec783130b018516d46115954565d9a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-11-12 10:52:49.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit ca0c58b52756058b6d51bf6a47aeac3d525c1e16","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c43f19e8456b9e20f03709773fb2ffdb94807a0a","revisions":{"51830e0f520d259c8075f448d5ac26072f2f70f6":{"kind":"REWORK","_number":1,"created":"2020-08-17 18:01:24.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/79/746579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/79/746579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/79/746579/1"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 15:35:03.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 17:54:40.000000000","tz":60},"subject":"Ramp up rbd resize to avoid excessive calls","message":"Ramp up rbd resize to avoid excessive calls\n\nChange the RBD store to resize the image by up to 8GiB\nat the time to not resize on every write.\n\nTrim the image in Ceph after all data has been written\nto the actual size in case we overshot the resize.\n\nPartial-Bug: #1792710\nRelated-to: spec-lite-Ceph-Store-Optimization\n\nChange-Id: I7f0bffda222b663d4316c5d6c03fdbd0d3337035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/51830e0f520d259c8075f448d5ac26072f2f70f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/51830e0f520d259c8075f448d5ac26072f2f70f6"}]},"branch":"refs/heads/master"},"2f7da62ef657d3009d06c72f79664a522f171fe5":{"kind":"REWORK","_number":2,"created":"2020-08-17 18:40:38.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/79/746579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/79/746579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/79/746579/2"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 15:35:03.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 18:40:10.000000000","tz":60},"subject":"Ramp up rbd resize to avoid excessive calls","message":"Ramp up rbd resize to avoid excessive calls\n\nChange the RBD store to resize the image by up to 8GiB\nat the time to not resize on every write.\n\nTrim the image in Ceph after all data has been written\nto the actual size in case we overshot the resize.\n\nPartial-Bug: #1792710\nRelated-to: spec-lite-Ceph-Store-Optimization\n\nChange-Id: I7f0bffda222b663d4316c5d6c03fdbd0d3337035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2f7da62ef657d3009d06c72f79664a522f171fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/2f7da62ef657d3009d06c72f79664a522f171fe5"}]},"branch":"refs/heads/master"},"3ef67618448ce7d796162d5587d464c98ba8e96b":{"kind":"REWORK","_number":3,"created":"2020-08-18 12:58:43.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/79/746579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/79/746579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/79/746579/3"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 15:35:03.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-18 12:58:29.000000000","tz":60},"subject":"Ramp up rbd resize to avoid excessive calls","message":"Ramp up rbd resize to avoid excessive calls\n\nChange the RBD store to resize the image by up to 8GiB\nat the time to not resize on every write.\n\nTrim the image in Ceph after all data has been written\nto the actual size in case we overshot the resize.\n\nPartial-Bug: #1792710\nRelated-to: spec-lite-Ceph-Store-Optimization\n\nChange-Id: I7f0bffda222b663d4316c5d6c03fdbd0d3337035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3ef67618448ce7d796162d5587d464c98ba8e96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3ef67618448ce7d796162d5587d464c98ba8e96b"}]},"branch":"refs/heads/master"},"6940e0dd5261edd81f7f4d55fa6361c36c736fdb":{"kind":"REWORK","_number":4,"created":"2020-08-18 16:02:41.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/79/746579/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/79/746579/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/79/746579/4"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 15:35:03.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-18 16:02:34.000000000","tz":0},"subject":"Ramp up rbd resize to avoid excessive calls","message":"Ramp up rbd resize to avoid excessive calls\n\nChange the RBD store to resize the image by up to 8GiB\nat the time to not resize on every write.\n\nTrim the image in Ceph after all data has been written\nto the actual size in case we overshot the resize.\n\nPartial-Bug: #1792710\nRelated-to: spec-lite-Ceph-Store-Optimization\n\nChange-Id: I7f0bffda222b663d4316c5d6c03fdbd0d3337035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6940e0dd5261edd81f7f4d55fa6361c36c736fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6940e0dd5261edd81f7f4d55fa6361c36c736fdb"}]},"branch":"refs/heads/master"},"a803988a880233c043b3736f2649ef248c03503d":{"kind":"REWORK","_number":5,"created":"2020-08-18 16:05:18.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/79/746579/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/79/746579/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/79/746579/5"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 15:35:03.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-18 16:05:13.000000000","tz":0},"subject":"Ramp up rbd resize to avoid excessive calls","message":"Ramp up rbd resize to avoid excessive calls\n\nChange the RBD store to resize the image by up to 8GiB\nat the time to not resize on every write.\n\nTrim the image in Ceph after all data has been written\nto the actual size in case we overshot the resize.\n\nPartial-Bug: #1792710\nRelated-to: spec-lite-Ceph-Store-Optimization\n\nChange-Id: I7f0bffda222b663d4316c5d6c03fdbd0d3337035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a803988a880233c043b3736f2649ef248c03503d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/a803988a880233c043b3736f2649ef248c03503d"}]},"branch":"refs/heads/master"},"c43f19e8456b9e20f03709773fb2ffdb94807a0a":{"kind":"NO_CHANGE","_number":6,"created":"2020-08-20 15:45:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/746579/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/79/746579/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/79/746579/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/79/746579/6"}}},"commit":{"parents":[{"commit":"3b71ce2d476a9fc4fa86091711d0b8ca1dff8868","subject":"Copy data files to glance upon installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/3b71ce2d476a9fc4fa86091711d0b8ca1dff8868"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-17 15:35:03.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-20 15:45:48.000000000","tz":-420},"subject":"Ramp up rbd resize to avoid excessive calls","message":"Ramp up rbd resize to avoid excessive calls\n\nChange the RBD store to resize the image by up to 8GiB\nat the time to not resize on every write.\n\nTrim the image in Ceph after all data has been written\nto the actual size in case we overshot the resize.\n\nPartial-Bug: #1792710\nRelated-to: spec-lite-Ceph-Store-Optimization\n\nChange-Id: I7f0bffda222b663d4316c5d6c03fdbd0d3337035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c43f19e8456b9e20f03709773fb2ffdb94807a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/c43f19e8456b9e20f03709773fb2ffdb94807a0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
