)]}'
{"id":"openstack%2Fglance_store~748723","triplet_id":"openstack%2Fglance_store~master~If5447c3f5b4f678943273c2bdb2e44cd25d5b226","project":"openstack/glance_store","branch":"master","topic":"drop-error-level","hashtags":[],"change_id":"If5447c3f5b4f678943273c2bdb2e44cd25d5b226","subject":"Drop snapshot in use log from ERROR to WARN","status":"MERGED","created":"2020-08-28 17:15:40.000000000","updated":"2020-08-30 15:14:35.000000000","submitted":"2020-08-28 20:40:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748723-1598647221868-33708db1","meta_rev_id":"316e5661d42ca6f98a59fa5798ee77c7307b64cd","_number":748723,"virtual_id_number":748723,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 20:40:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-28 18:14:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-28 17:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-28 18:14:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 17:37:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-08-28 18:14:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-28 20:40:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c107a77f584a1e8e5f1f3e7608cad979b09da69","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-28 17:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d09470c444b03f4bc8221681478f8bd203aeb5d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-28 17:20:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"59dec082f00ec7b4f962b45e794630f748212706","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-28 17:26:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f6a365c6ae5931a373d619323ab55fbfe68626af","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-28 17:37:07.000000000","message":"Patch Set 3: Code-Review+2\n\nSeems okay to me. This exception is caught and logged as warning on the glance side already, so logging it here as error and there again as warning makes little sense.","accounts_in_message":[],"_revision_number":3},{"id":"4a513dba46687f253dabe6108888ead7bd28ce7b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-28 18:14:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nYep, makes sense.","accounts_in_message":[],"_revision_number":3},{"id":"3d530047ec36a8686013b0631a14a5a3c5ef195a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:07:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1348c3cafbad4565ad943521e019b08e : SUCCESS in 1h 18m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3744af5279ae46ca9976e5c8cd7de58b : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f59703cea04fc9b751e740650910ec : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f04385315f404bc68406baebed7377c3 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06d8099c5bc3403f8309d954e048d31d : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8efb0ca389d401c8c67e4c45d903732 : SUCCESS in 4m 32s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e3d8a199c1e4cf7ae59b8594bf13dc2 : SUCCESS in 1h 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"86ffa98931bfcb5867d321aedc8ae218ba829762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:07:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5099a10d6f9982ade4934a2fc5394840d8d8f679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 20:40:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6e1d535d5416f939aba326cee87b4e3071858d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 20:40:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/416d66eef0c046f79bfb7e46797561f1 : SUCCESS in 1h 30m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b123f5ba7d946d7964524a4656c743e : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e934a5d1cd460b8b7f858fde989b32 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1eaeef9de59424ca39541a7f399e47b : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a44e3105b464361a2317450e498fe65 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf78eeac1f4e4c929a8510891c34ff8e : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"fe88f72dfca55e6186f4a92f7fcb0fefa0f859c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 20:42:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20b6485327349ac9213d7c1dfaef110 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cea994baf70a3db5b9092bf005a5c7f8852747a7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-30 15:14:35.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 3914372d259857d3622ffb8b63ecc22e7bd45ef3","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5aa2bc7af6f22776ea5751b7f9620845bb5f3a3a","revisions":{"ad5dcfca9b4b79fc441337b8244bc5b99d44d3b7":{"kind":"REWORK","_number":1,"created":"2020-08-28 17:15:40.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/23/748723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/23/748723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/23/748723/1"}}},"commit":{"parents":[{"commit":"f076e7e4269868ba4d9c10e6eeda0f84ce2552c6","subject":"Merge \"Clarify the filesystem_store_metadata_file config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f076e7e4269868ba4d9c10e6eeda0f84ce2552c6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-08-28 17:14:19.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-08-28 17:15:38.000000000","tz":-240},"subject":"Drop snapshot in use log from ERROR to WARN","message":"Drop snapshot in use log from ERROR to WARN\n\nThis message is not actually an error inside the system for most of the\ntime.  The primary reason behind this is that it is possible that there\nare VMs using this image and therefore it can\u0027t be deleted, the operator\nhas no actionable thing to do about this.\n\nFor those who are monitoring ERROR level messages and alerting on them,\nthis is noisy and not actionable, drop this error message to warning.\n\nChange-Id: If5447c3f5b4f678943273c2bdb2e44cd25d5b226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ad5dcfca9b4b79fc441337b8244bc5b99d44d3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/ad5dcfca9b4b79fc441337b8244bc5b99d44d3b7"}]},"branch":"refs/heads/master"},"135a97a71a7351acbd44f0d9619598bdb06a47c2":{"kind":"REWORK","_number":2,"created":"2020-08-28 17:20:34.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/23/748723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/23/748723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/23/748723/2"}}},"commit":{"parents":[{"commit":"f076e7e4269868ba4d9c10e6eeda0f84ce2552c6","subject":"Merge \"Clarify the filesystem_store_metadata_file config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f076e7e4269868ba4d9c10e6eeda0f84ce2552c6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-08-28 17:14:19.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-08-28 17:20:29.000000000","tz":-240},"subject":"Drop snapshot in use log from ERROR to WARN","message":"Drop snapshot in use log from ERROR to WARN\n\nThis message is not actually an error inside the system for most of the\ntime.  The primary reason behind this is that it is possible that there\nare VMs using this image and therefore it can\u0027t be deleted, the operator\nhas no actionable thing to do about this.\n\nFor those who are monitoring ERROR level messages and alerting on them,\nthis is noisy and not actionable, drop this error message to warning.\n\nChange-Id: If5447c3f5b4f678943273c2bdb2e44cd25d5b226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/135a97a71a7351acbd44f0d9619598bdb06a47c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/135a97a71a7351acbd44f0d9619598bdb06a47c2"}]},"branch":"refs/heads/master"},"5aa2bc7af6f22776ea5751b7f9620845bb5f3a3a":{"kind":"REWORK","_number":3,"created":"2020-08-28 17:26:49.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/23/748723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/23/748723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/23/748723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/23/748723/3"}}},"commit":{"parents":[{"commit":"f076e7e4269868ba4d9c10e6eeda0f84ce2552c6","subject":"Merge \"Clarify the filesystem_store_metadata_file config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/f076e7e4269868ba4d9c10e6eeda0f84ce2552c6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-08-28 17:14:19.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-08-28 17:26:43.000000000","tz":-240},"subject":"Drop snapshot in use log from ERROR to WARN","message":"Drop snapshot in use log from ERROR to WARN\n\nThis message is not actually an error inside the system for most of the\ntime.  The primary reason behind this is that it is possible that there\nare VMs using this image and therefore it can\u0027t be deleted, the operator\nhas no actionable thing to do about this.\n\nFor those who are monitoring ERROR level messages and alerting on them,\nthis is noisy and not actionable, drop this error message to warning.\n\nChange-Id: If5447c3f5b4f678943273c2bdb2e44cd25d5b226\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5aa2bc7af6f22776ea5751b7f9620845bb5f3a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5aa2bc7af6f22776ea5751b7f9620845bb5f3a3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
