)]}'
{"id":"openstack%2Fglance_store~769682","triplet_id":"openstack%2Fglance_store~master~Ifa78924d7ecf4f2d9a54c677888ab2926530c487","project":"openstack/glance_store","branch":"master","topic":"remove-py2-support","attention_set":{},"removed_from_attention_set":{"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2022-07-07 17:23:38.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-07 17:23:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa78924d7ecf4f2d9a54c677888ab2926530c487","subject":"Remove Python 2 support","status":"MERGED","created":"2021-01-07 07:58:09.000000000","updated":"2022-07-07 17:24:42.000000000","submitted":"2022-07-07 17:23:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769682-remove-py2-support","meta_rev_id":"78da60fc77b1ccba86267a8f8331ef296ae2ecff","_number":769682,"virtual_id_number":769682,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 17:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-07-07 15:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-07-07 15:48:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 08:06:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 12:45:18.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-07-07 15:48:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"57c673baac4dd3cf8ddf80a40b5e9928315777d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2021-01-07 07:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48b1ccd6ea2884561fb762ff284ef5867e854b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 08:06:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/901d0f49c753443492edaa0263b44499 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d65e67d1814de79979b936b193a894 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcd16174a53a44c09fe598a0ed368f76 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7a035891bbf0b2de6c06d0167ef83d419603a15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-12-15 02:58:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56f30122cac3f941aea1ce4eaa59576353d01d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 04:42:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/65d008c8e95047aba8882cc8fa678bc3 : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f245c54d9754492b91e2e17941b3976 : FAILURE in 1h 02m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395a614430694da8b506077de3f582ea : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2454404a743841f2a43b2b5b933154d8 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49fb383ce9f54f08afab26a13eaf8250 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf69f636b3a040b4b92fe8d0d2570ee3 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c3b9eb7f25f48b0b6bbe79c57fd6877 : SUCCESS in 4m 28s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/b3b0cdd3cf7f4d0d891086b77cb8867e : SUCCESS in 11m 23s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/4278e51148484f7b9e6677fadd23e04d : SUCCESS in 1h 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a679bb276d6c387aa02104ef15043f3b46201aca","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-12-15 12:45:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"025216bcedeeff5cd1bf5c5d5b7d0a0d7c9e060d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 14:26:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/263c330bf25b4478a55a65f1e3a292be : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30e080a0f1b0417fb515f7bbaa881cb2 : SUCCESS in 1h 27m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f05e5d889c433a99a5d24b04bdf495 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/703d6ce4963847c5aa3de8b71d908141 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/523806c726d4463b88a1e1be4913edcf : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74de377cd1624e2d9a971cea77be5ec3 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4828f65d974b42acac672a0d50de55c8 : SUCCESS in 3m 31s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/2ffcb749e1c04e6e965701ff58667adc : SUCCESS in 8m 41s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/534dc78bc0084b90a87f77ccb264c5e1 : SUCCESS in 1h 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9672419c2b3201d69f1638449c6b680feff09e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-12-15 20:31:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a88c7d6744f0cc35916e5abc9b8cd1479a6bd9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 21:55:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/439f40e3ce3c4b0d9a593203ef8b7f3b : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b80b01a6c41c468584912c0a6f8793cb : SUCCESS in 1h 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf9b8980d6648b0a7e4e617ad23a3fa : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbe35de92b484a008e2b2fcefa3e220d : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed7c67a33246463f9ad2a1d770864290 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05965f6a58a43cd8a7c6c269e1e0f2b : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08d25d24bb6040ea9e48ebfda07b679c : SUCCESS in 4m 35s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f12fe1a8479e421581959a9f5dfc10ba : SUCCESS in 8m 17s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/97addc5f9ba6421199dd93cdc09ad4e9 : FAILURE in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8deb255986141d56625ee40c548bae014d7086c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-25 02:48:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0976f98e1d6d87453ca5b164b3c386ea451899bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 05:06:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c125721112240d98e00b304466288ae : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fde31eaadb34e7bb7f5ffc439eaf555 : SUCCESS in 1h 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d6aeee140e4f1184598b9b84b8866c : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c4de968040945e7a6b3d0b800f4550b : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca435a26ea25435f9bc87756b5838520 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/594b6bdd76f548129608ebbf48ba3b46 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd18668a0e448eeb2e19fa8f64c073b : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/413cf6514e0f450581699aeddc591d78 : SUCCESS in 4m 18s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/ab83f5d14c5e4a8082a192a5442a2a9b : SUCCESS in 8m 29s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a413728be3241b5864f832087728fb8 : TIMED_OUT in 2h 04m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc96ead0b0417b65b2ea625db7c9419aa0e9de2d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 15:48:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"77dcee8ae5d0aa55dc99ea07788f9c6b825a23ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:48:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5f876030fff088dd78a74f950173f02f38afdc90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 17:23:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7477d661ba4d94bfce7d9c8e47c7a1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c61704f85c0842d7a61e818546fc6e77 : SUCCESS in 2m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5a3252a7ca54ef8a038673d08d94442 : SUCCESS in 1h 34m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8b70deedac4cb9b077583f9d34a877 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c64e03b3d9e8475faa8147632ac0674a : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b39d87a0ff4c14b15329b970b235b0 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa6adadaccb4d9a86d9e1fa3fb7a57d : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54ecc9d4d82243f088175511c05ec299 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":4},{"id":"89146c5a4a76640874cf03fac00e19ca5461fbba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 17:23:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"78da60fc77b1ccba86267a8f8331ef296ae2ecff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 17:24:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/067bab5940634afea839ed38b91654c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2ae6c59e87429f8e481c70a70fdf2c : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81e81469162f4c5e82d88867e5641be7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5ff06df97fbd78f6781793c35801a4c989a515f1","revisions":{"9ca3c7aeb7818793fc3d22dc48c292a857005806":{"kind":"REWORK","_number":1,"created":"2021-01-07 07:58:09.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/82/769682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/769682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/769682/1"}}},"commit":{"parents":[{"commit":"d05024afc3bed9deca1acc49f4a8d6faf14b0545","subject":"Fix lower_constraints and requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/d05024afc3bed9deca1acc49f4a8d6faf14b0545"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-01-07 07:57:43.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-01-07 07:58:02.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Ifa78924d7ecf4f2d9a54c677888ab2926530c487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9ca3c7aeb7818793fc3d22dc48c292a857005806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/9ca3c7aeb7818793fc3d22dc48c292a857005806"}]},"branch":"refs/heads/master"},"30c7024088d176ee7eb580253e6a17471ce92b58":{"kind":"REWORK","_number":2,"created":"2021-12-15 02:58:33.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/769682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/769682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/769682/2"}}},"commit":{"parents":[{"commit":"6295d2eead3e492959f5861ded6a6c3e3f8cc9d0","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6295d2eead3e492959f5861ded6a6c3e3f8cc9d0"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-01-07 07:57:43.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-12-15 02:44:39.000000000","tz":60},"subject":"remove unicode from code","message":"remove unicode from code\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ifa78924d7ecf4f2d9a54c677888ab2926530c487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/30c7024088d176ee7eb580253e6a17471ce92b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/30c7024088d176ee7eb580253e6a17471ce92b58"}]},"branch":"refs/heads/master"},"aa654e28b3696c01c9fe549518902d71b98e3644":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-12-15 20:31:57.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/769682/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/769682/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/769682/3"}}},"commit":{"parents":[{"commit":"6295d2eead3e492959f5861ded6a6c3e3f8cc9d0","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/6295d2eead3e492959f5861ded6a6c3e3f8cc9d0"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-01-07 07:57:43.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-12-15 20:30:49.000000000","tz":60},"subject":"Remove Python 2 support","message":"Remove Python 2 support\n\nPython 2 has been deprecated for almost two years, and has not been\nguaranteed to work with glance_store for a while. This patch removes all\ntraces of six, unicode strings and Python 2 tweaks.\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ifa78924d7ecf4f2d9a54c677888ab2926530c487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aa654e28b3696c01c9fe549518902d71b98e3644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aa654e28b3696c01c9fe549518902d71b98e3644"}]},"branch":"refs/heads/master"},"5ff06df97fbd78f6781793c35801a4c989a515f1":{"kind":"REWORK","_number":4,"created":"2022-05-25 02:48:31.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/82/769682/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/82/769682/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/82/769682/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/82/769682/4"}}},"commit":{"parents":[{"commit":"aeee48b561ba0d100663311c3ba94f0d128105aa","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/aeee48b561ba0d100663311c3ba94f0d128105aa"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2021-01-07 07:57:43.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-05-25 02:42:06.000000000","tz":120},"subject":"Remove Python 2 support","message":"Remove Python 2 support\n\nPython 2 has been deprecated for almost two years, and has not been\nguaranteed to work with glance_store for a while. This patch removes all\ntraces of six, unicode strings and Python 2 tweaks.\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ifa78924d7ecf4f2d9a54c677888ab2926530c487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5ff06df97fbd78f6781793c35801a4c989a515f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/5ff06df97fbd78f6781793c35801a4c989a515f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
