)]}'
{"id":"openstack%2Fglance_store~776502","triplet_id":"openstack%2Fglance_store~master~Icf96eb49d560f2deb543232435094b34934b9406","project":"openstack/glance_store","branch":"master","topic":"cross-glance-functional","hashtags":[],"change_id":"Icf96eb49d560f2deb543232435094b34934b9406","subject":"Run glance functional job on glance_store","status":"MERGED","created":"2021-02-18 17:15:29.000000000","updated":"2021-02-19 22:44:28.000000000","submitted":"2021-02-19 22:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"776502-cross-glance-functional","meta_rev_id":"d8302db58d10e3dd5f4aec87e7db353031f25aab","_number":776502,"virtual_id_number":776502,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 22:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-02-19 16:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-02-19 16:50:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 17:37:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-02-18 17:37:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-02-18 17:37:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-18 17:37:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-18 19:05:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33a5c946a0e4979cdd71d57749ef79748c007aaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-18 17:15:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee561b63daaf4e01553183f897e8d9eddd789e94","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-18 17:37:07.000000000","message":"Patch Set 1:\n\nAs discussed on the meeting.\n\nThis properly ran glance functional against this glance_store patch and poked the assertion bug that Rajat is fixing:\n\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_78d/776502/1/check/cross-glance-tox-functional/78df6b1/testr_results.html\n\nSo after that fix merges, we can land this and next time we will see it early.","accounts_in_message":[],"_revision_number":1},{"id":"3a6e0a3b2b92bd04eda1686c0db688cd23aa3d7f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-18 17:49:11.000000000","message":"Patch Set 1: Code-Review+1\n\nFails[1] on the test noticed in the tips job\nI feel this is a good addition to cross check any change in glance_store master not breaking glance later (after release)\nHoping this is enough compensation for the time/resources taken for this job run, LGTM.\nThanks Dan.\n\n[1] https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_78d/776502/1/check/cross-glance-tox-functional/78df6b1/testr_results.html","accounts_in_message":[],"_revision_number":1},{"id":"dbedca955453a22e257ee2f95f640eab86a38a87","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-18 17:50:11.000000000","message":"Patch Set 1:\n\na recheck after change[1] merges should pass this job\n\n[1] https://review.opendev.org/c/openstack/glance/+/776369","accounts_in_message":[],"_revision_number":1},{"id":"afbda7abcce17238d6665d9b4b955dd6b1c309ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 19:05:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3331b0f5215f484391a1c3bd157b2b68 : SUCCESS in 1h 28m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1197449ee29440e390f2cf1b9edb8edf : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8611ca9a7b724ca2b07d1f36bd399784 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/146c834aaeda49189b4bb581f79a9a7f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3207bf46d53e41edb38ba49867bbecef : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81fe0f184845460e871301a4ad0bf277 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8cd1b6725ce4d2fab298336b38d5865 : SUCCESS in 4m 13s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/78df6b18b1c642248cd0b55270a11fb8 : FAILURE in 7m 55s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce165d4bfb644673adb15192c40a8e58 : SUCCESS in 1h 47m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bee89ca2a7f417a31ca9b54e8c534bb9746bbc95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-18 19:22:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a03e9b410b926c7a084a6ef49157d90de2b2c154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:07:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b65fe0ff9a6644c8806854045a927065 : SUCCESS in 1h 18m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cedd7997def4d19ad3a6c687af5d7ac : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea25771a10e4a15b2156af5c83fdc74 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f3cde3a6b364629a2a9a087b4a516e4 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5621e2d347d34eef85a925b88431e89e : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d889c7542adc43fd94c0b87b650021a6 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cfec523f6ab46e0901a2c1570aee41e : SUCCESS in 4m 28s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/f4d56a158f214ad2995649c3c067b7e3 : SUCCESS in 13m 30s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/d41ef8736b9149839b75cf7c30c8fcd0 : SUCCESS in 1h 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f851a9dc3063439e4191ec3c71224973124a8ba8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-18 21:10:01.000000000","message":"Patch Set 2:\n\n...and passes now that this depends-on the fix.","accounts_in_message":[],"_revision_number":2},{"id":"32b31c05b0d739a2de971a17ea6ced2eb707d948","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 14:31:09.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nThis will definitely help in future.\nThank you Dan!","accounts_in_message":[],"_revision_number":2},{"id":"c5a300fab5981d46ca33155c577ffa721969d2ad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-19 16:44:35.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nNit noted inline.","accounts_in_message":[],"_revision_number":2},{"id":"e87caa4b55ef8188a1e89ab7bac23724bd8752ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 16:45:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"295cabc0dd4977d841a507f637fde0d5a7604f0b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-19 16:50:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nChange was to a comment, so carrying over Abhishek\u0027s +2.  LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"747b10c41ef3306f759126a55eea858c36b80530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:26:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2864c71a6e7e4b0fab60c904eb6c070f : POST_FAILURE in 1h 23m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a788cc1200fd448c91470222961ed409 : POST_FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb8fa7637b04021a8d2cccbe887cea4 : POST_FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94bf9e6a02384d529545fb5c3eb9f78b : POST_FAILURE in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/036b245d590547acb202068648b46c87 : POST_FAILURE in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0599868dae1c48989b40e8375c663dc3 : POST_FAILURE in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a27ec1e586e42b7b7545834fffca829 : POST_FAILURE in 5m 50s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/c5780bb5252649edb88053934b81de52 : POST_FAILURE in 8m 52s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ecab464e74a4bc08421ca69e60230da : POST_FAILURE in 1h 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3931f9c794466d233e8c7f756ba2f682f9553296","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 18:53:25.000000000","message":"Patch Set 3:\n\nrecheck known infra failure","accounts_in_message":[],"_revision_number":3},{"id":"3c59f983c5549ebfad582196682866629cffdf04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:58:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2209d03da424da6ab91794a9f9e6a89 : SUCCESS in 1h 47m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b2c8846b6dc45cfa03d03b6e9748d88 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c601d06007ba42eeae74bb76c4d09231 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fadf8de2c7a444e4b803c036293798d0 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88809196e4fa41dead10ddab838fb3df : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df14797af5974895b60a1924368a3eda : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469216880a714df98a1954865ad77f9d : SUCCESS in 3m 53s\n- cross-glance-tox-functional https://zuul.opendev.org/t/openstack/build/3339b3d47eca47b8ba0723356c42a608 : SUCCESS in 8m 02s\n- glance_store-src-ceph-tempest https://zuul.opendev.org/t/openstack/build/432978304c3d4eeca477f9f607d8edda : SUCCESS in 1h 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26749e045995824095752d619b61b5f9052a8035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:59:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de2019b7df66d8542f5b8fad534d368b170ebef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:38:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a87ea2c2be5242a2ac1e3c7b6edb49a3 : SUCCESS in 1h 36m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1714736ddda4264af1dd67617eefdd7 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fce06b1d4034c50a7c96b67862ee4e4 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df8c7e88fcbf46ccaff72537bea49927 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81ca736dbf5e4f9a94f0912c39f472cf : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8eca661bbe34d2da65932703077afea : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"4e4bcbad560233e34eba0d418f7b0dba55dafa0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:38:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d8302db58d10e3dd5f4aec87e7db353031f25aab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:44:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e0b152dc0b4866a7337acf9ae9bc76 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18dd4421e46cad6ecd03b611b2e9b50c0e4461de","revisions":{"848b6640f3738ffbc99655acc29729f38228ab59":{"kind":"REWORK","_number":1,"created":"2021-02-18 17:15:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/776502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/02/776502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/02/776502/1"}}},"commit":{"parents":[{"commit":"00025156e0963d9e4752a1d45223aeca1dea73de","subject":"Merge \"Validate volume type during volume create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/00025156e0963d9e4752a1d45223aeca1dea73de"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-18 18:12:37.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-18 18:12:37.000000000","tz":-480},"subject":"Run glance functional job on glance_store","message":"Run glance functional job on glance_store\n\nThis will let us see if a proposed patch against glance_store will\nbreak glance\u0027s functional job. We already run a tempest job against\nglance_store so we should be loosely covered there already.\n\nChange-Id: Icf96eb49d560f2deb543232435094b34934b9406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/848b6640f3738ffbc99655acc29729f38228ab59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/848b6640f3738ffbc99655acc29729f38228ab59"}]},"branch":"refs/heads/master"},"cece7fe4c08cef98fb3b100c90ae2af718316a15":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-18 19:22:59.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/776502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/02/776502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/02/776502/2"}}},"commit":{"parents":[{"commit":"00025156e0963d9e4752a1d45223aeca1dea73de","subject":"Merge \"Validate volume type during volume create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/00025156e0963d9e4752a1d45223aeca1dea73de"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-18 18:12:37.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-18 20:21:24.000000000","tz":-480},"subject":"Run glance functional job on glance_store","message":"Run glance functional job on glance_store\n\nThis will let us see if a proposed patch against glance_store will\nbreak glance\u0027s functional job. We already run a tempest job against\nglance_store so we should be loosely covered there already.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/776369\nChange-Id: Icf96eb49d560f2deb543232435094b34934b9406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/cece7fe4c08cef98fb3b100c90ae2af718316a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/cece7fe4c08cef98fb3b100c90ae2af718316a15"}]},"branch":"refs/heads/master"},"18dd4421e46cad6ecd03b611b2e9b50c0e4461de":{"kind":"REWORK","_number":3,"created":"2021-02-19 16:45:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/776502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance_store","ref":"refs/changes/02/776502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance_store refs/changes/02/776502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance_store refs/changes/02/776502/3"}}},"commit":{"parents":[{"commit":"00025156e0963d9e4752a1d45223aeca1dea73de","subject":"Merge \"Validate volume type during volume create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/00025156e0963d9e4752a1d45223aeca1dea73de"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-18 18:12:37.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 17:44:41.000000000","tz":-480},"subject":"Run glance functional job on glance_store","message":"Run glance functional job on glance_store\n\nThis will let us see if a proposed patch against glance_store will\nbreak glance\u0027s functional job. We already run a tempest job against\nglance_store so we should be loosely covered there already.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/776369\nChange-Id: Icf96eb49d560f2deb543232435094b34934b9406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/18dd4421e46cad6ecd03b611b2e9b50c0e4461de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance_store/commit/18dd4421e46cad6ecd03b611b2e9b50c0e4461de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
